)]}'
{"id":"openstack%2Fproject-config~868526","triplet_id":"openstack%2Fproject-config~master~I05d36dab6895dfc14d026230592def51b78bb755","project":"openstack/project-config","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-03 10:52:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05d36dab6895dfc14d026230592def51b78bb755","subject":"Fix tox4 error","status":"MERGED","created":"2022-12-25 02:55:30.000000000","updated":"2023-01-03 10:53:46.000000000","submitted":"2023-01-03 10:52:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868526-tox4","meta_rev_id":"62db21cb388fa2c6dda79a430e250cb03d36e50a","_number":868526,"virtual_id_number":868526,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-03 10:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-01-03 10:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-25 13:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-01-03 10:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-25 03:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-25 13:51:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-01-03 10:43:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9952d8e627312bd354bd8d1263ef441b7074115e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 02:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7347abdd69d8c2e2bbfa90bf9c4503b4be5b757b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 02:56:31.000000000","message":"Topic set to tox4","accounts_in_message":[],"_revision_number":1},{"id":"05d32cb8d965bfef70a102f8e6f6e8d05906bf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 03:05:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1711fde8075d4378b09a215de5022377\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f62c370eca3d4721a1e18fbf13214c37 : SUCCESS in 4m 30s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/d852ac4a46fe408da4e693a84f2cf0bc : FAILURE in 3m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/061072db774f47e9a5082759c61ad73f : FAILURE in 4m 41s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/958ebf576c4a41cabcd9efd72af2bf7e : FAILURE in 3m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a1c116e58f1e4415a3e934ee6524da6f : FAILURE in 5m 04s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/5f4321779d1f4aa986029efe0b6a0810 : FAILURE in 3m 48s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/aa4d5979fd064a2daaaed80cadc501a9 : FAILURE in 4m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/25124cfe457440d8a64da1157823c618 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3c124065b8010593fc755a0639e9377ac359e7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 03:11:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cff18d86d337e4420c42429f034a51ec6e6df23e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 03:26:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3c6605c3ef34614ba952dbdb54883a2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/588577d2f68443f6afdc07cd6907a5a6 : SUCCESS in 3m 19s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/941229aeb728480784f175f11cc2f846 : FAILURE in 3m 14s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7bbf05bd42b6465e99accd26760921c6 : SUCCESS in 7m 22s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/8b355c21a19d4debb17dfac79bf7839d : FAILURE in 3m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9b3cc49b4a2e4c80a723639d14b73c05 : FAILURE in 3m 22s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/266de8ea6ba14e43af5ff9a02d28d1ed : FAILURE in 3m 28s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1096011b45c74ea296049b30738e04c7 : FAILURE in 4m 03s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d5b3bd5071c44432a596095e3a24543a : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8bfe06e676fa77bca829fb5b742c959f68b182bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 03:35:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2ebf787793dd2490b23120ae8ca04d94360af6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 03:46:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29654ce18a9b49a4b6b66978ddfe64af\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/22c594c952c54319a011850196e89440 : SUCCESS in 3m 04s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/dad94cb486d0420b868f1da014f037ae : FAILURE in 2m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5cc48728289a4999940905a4437d4f07 : FAILURE in 3m 02s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/77c898bdf8ce4ec8888b0279e5a29513 : FAILURE in 2m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/12ccb585d3dd4c08a7f40bbbf78fa307 : FAILURE in 2m 55s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/8524467b3461462eb1a150fee118b453 : FAILURE in 2m 51s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4ba645112c034eeca32214c4b099e742 : FAILURE in 2m 56s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4ffce016ed5a42f7b0d3d3ce93611615 : FAILURE in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"56bc1f055794ffe7c09bd73d08829c089e12b427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 03:48:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9dd32da57fcd27aaaa4b1197b8f2c8eca1ce1542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 04:02:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15c9fd62623845c78f13f3087daebe9f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/97803c1890b24d829bdd463143e9e0e2 : SUCCESS in 2m 39s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/56ad851004174b2788b7de7c9a6f964a : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a39c689fc2f44a9698b87b014567374f : SUCCESS in 6m 44s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6ec2da53760a4f8a94ce05ee88acfcce : FAILURE in 3m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a074f2f278bb474ebe01e90d39908a5c : SUCCESS in 3m 32s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/e6886376e9074705b0a7151757793fb6 : SUCCESS in 3m 16s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/eb446c60fc2e46279ceda561d336b40d : FAILURE in 5m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/fab551234a7648d7a78618cdf6f33f35 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"bf72543c5eae022d9d5780a29f035f3ee2f69ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 04:20:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d20d1a9af3b19a8a5c0351a33ba5b51492047938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 04:32:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b91f40bd81d0459eb2cb2bb770e7977e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ce5533d481164925be509381a459427a : SUCCESS in 3m 10s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/420ccdf4189c489dbd3c8708dc5e24a5 : SUCCESS in 3m 07s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4cd7e49f8fda4b5a9a37d7eedabd6dc4 : SUCCESS in 6m 30s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/b48190c1f4d34073871b24d986728890 : SUCCESS in 3m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c6b45de1c54643468f56427c1d6d7a7c : SUCCESS in 3m 41s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/347feea7edee4ce79f97d56fd8bb0872 : SUCCESS in 3m 09s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8ea91583d56b42e68311252a926e01fc : POST_FAILURE in 6m 02s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1a66014064794af98053aa9a3ac4ebc5 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"7a441472193358031dbcf0e480317e9ad01e2895","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-25 04:33:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"979fd5db9ebcd666ec74e578340e873fc6d9e621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-25 04:44:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e62448e7e5c84257a7a466e4042fb430\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e089161562c440af8fcfb5e3af823840 : SUCCESS in 2m 50s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/ed724feffe0247789ab511769c5c13f8 : SUCCESS in 3m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/329bb01e1bee409d9ab462930c1852a5 : SUCCESS in 6m 43s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/891bd60b668e4c6ebc4fad60505dba0f : SUCCESS in 3m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/925892fda60b420da20e6b24c13e5cb8 : SUCCESS in 3m 46s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/2df04786f2d7435093c156b25e551a8e : SUCCESS in 3m 18s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1a05af8241eb4dfc8102bba7c52f9244 : SUCCESS in 5m 30s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d0136cc128ce4d979ae79525c0aa469b : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"2468d1a8a0716f452057d33a6bf58014f9fdbbce","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-25 13:51:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"042c32155c2b7a26e94da6c7c0c5c25312b62eef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-03 10:43:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"294d722a96da1a412f4f37928edae2c4f5f246db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:44:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8c2c319af01c3a6037a13fbfc7ada5171f0d33d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:52:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72c33f32d7044b5abb27cf461f9a50c\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/14bc3581c1634ef3b89df5d9583d8b53 : SUCCESS in 3m 14s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d686b0496fa0418898f43fdb890950ed : SUCCESS in 6m 11s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/fc5b7d6ce7e240628a3ab2eb0e1656b1 : SUCCESS in 2m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/775772822ce14d71ae8f4efb568b607d : SUCCESS in 3m 35s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/8b0c1fa8f8ea4a7daa3cc649378606e9 : SUCCESS in 2m 50s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e55f2893f48e4c0081e9d46e314f75c7 : SUCCESS in 5m 25s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/f902bec13efd4671aabeafbad149538c : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"889c6ee9c1dfdd96d20d604f34bfb732556388db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:52:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"62db21cb388fa2c6dda79a430e250cb03d36e50a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 10:53:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5becac3f807c4477a8a1b57ba26e3268\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/0e9b446d86994881a5c2e75355259888 : SUCCESS in 44s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/f80bc6887d3442e09d4ea9cc6f4c3a8e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6facccc00e89c1176f35ee7a40af0743e8572a20","revisions":{"4e9e6acad87f235ceddd9f976209c97faaff8ffc":{"kind":"REWORK","_number":1,"created":"2022-12-25 02:55:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/868526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/868526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/868526/1"}}},"commit":{"parents":[{"commit":"4bc39369453896d4fa52ecd80244ae2134657497","subject":"Update Grafana\u0027s dasboard for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39369453896d4fa52ecd80244ae2134657497"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 02:51:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 02:51:03.000000000","tz":-480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\nWith tox4 in use for testing, tox\nrun failing with pass_env value.\n\nfailed with pass_env values cannot contain\nwhitespace, use comma to have multiple values in a single\nline, invalid values found \u0027http_proxy HTTP_PROXY https_proxy\nHTTPS_PROXY no_proxy NO_PROXY\u0027 [tox/session/cmd/run/single.py:54]\n\nChange-Id: I05d36dab6895dfc14d026230592def51b78bb755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e9e6acad87f235ceddd9f976209c97faaff8ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e9e6acad87f235ceddd9f976209c97faaff8ffc"}]},"branch":"refs/heads/master"},"94a854c525ccb259c97f784385fed895e74f4943":{"kind":"REWORK","_number":2,"created":"2022-12-25 03:11:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/868526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/868526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/868526/2"}}},"commit":{"parents":[{"commit":"4bc39369453896d4fa52ecd80244ae2134657497","subject":"Update Grafana\u0027s dasboard for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39369453896d4fa52ecd80244ae2134657497"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 02:51:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:11:26.000000000","tz":0},"subject":"Fix tox4 error","message":"Fix tox4 error\n\nWith tox4 in use for testing, tox\nrun failing with pass_env value.\n\nfailed with pass_env values cannot contain\nwhitespace, use comma to have multiple values in a single\nline, invalid values found \u0027http_proxy HTTP_PROXY https_proxy\nHTTPS_PROXY no_proxy NO_PROXY\u0027 [tox/session/cmd/run/single.py:54]\n\nChange-Id: I05d36dab6895dfc14d026230592def51b78bb755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94a854c525ccb259c97f784385fed895e74f4943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94a854c525ccb259c97f784385fed895e74f4943"}]},"branch":"refs/heads/master"},"8f0d7bdf86457f21454d268aaf70cf27742923b5":{"kind":"REWORK","_number":3,"created":"2022-12-25 03:35:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/868526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/868526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/868526/3"}}},"commit":{"parents":[{"commit":"4bc39369453896d4fa52ecd80244ae2134657497","subject":"Update Grafana\u0027s dasboard for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39369453896d4fa52ecd80244ae2134657497"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 02:51:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:34:52.000000000","tz":-480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\nWith tox4 in use for testing, tox\nrun failing with passenv value and\nexternal commands run.\n\nfailed with pass_env values cannot contain\nwhitespace, use comma to have multiple values in a single\nline, invalid values found \u0027http_proxy HTTP_PROXY https_proxy\nHTTPS_PROXY no_proxy NO_PROXY\u0027 [tox/session/cmd/run/single.py:54]\n\nFixing passenv value and adding external commands\nvia allowlist_externals.\n\nChange-Id: I05d36dab6895dfc14d026230592def51b78bb755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f0d7bdf86457f21454d268aaf70cf27742923b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f0d7bdf86457f21454d268aaf70cf27742923b5"}]},"branch":"refs/heads/master"},"f38d09a7c57e5845db10a55795473908ae47dc7e":{"kind":"REWORK","_number":4,"created":"2022-12-25 03:48:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/868526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/868526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/868526/4"}}},"commit":{"parents":[{"commit":"4bc39369453896d4fa52ecd80244ae2134657497","subject":"Update Grafana\u0027s dasboard for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39369453896d4fa52ecd80244ae2134657497"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 02:51:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 03:48:09.000000000","tz":-480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\nWith tox4 in use for testing, tox\nrun failing with passenv value and\nexternal commands run.\n\nfailed with pass_env values cannot contain\nwhitespace, use comma to have multiple values in a single\nline, invalid values found \u0027http_proxy HTTP_PROXY https_proxy\nHTTPS_PROXY no_proxy NO_PROXY\u0027 [tox/session/cmd/run/single.py:54]\n\nFixing passenv value and adding external commands\nvia allowlist_externals.\n\nChange-Id: I05d36dab6895dfc14d026230592def51b78bb755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f38d09a7c57e5845db10a55795473908ae47dc7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f38d09a7c57e5845db10a55795473908ae47dc7e"}]},"branch":"refs/heads/master"},"6facccc00e89c1176f35ee7a40af0743e8572a20":{"kind":"REWORK","_number":5,"created":"2022-12-25 04:20:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/26/868526/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/868526/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/868526/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/868526/5"}}},"commit":{"parents":[{"commit":"4bc39369453896d4fa52ecd80244ae2134657497","subject":"Update Grafana\u0027s dasboard for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39369453896d4fa52ecd80244ae2134657497"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 02:51:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-25 04:20:21.000000000","tz":-480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\nWith tox4 in use for testing, tox\nrun failing with passenv value and\nexternal commands run.\n\nfailed with pass_env values cannot contain\nwhitespace, use comma to have multiple values in a single\nline, invalid values found \u0027http_proxy HTTP_PROXY https_proxy\nHTTPS_PROXY no_proxy NO_PROXY\u0027 [tox/session/cmd/run/single.py:54]\n\nFixing passenv value and adding external commands\nvia allowlist_externals.\n\nChange-Id: I05d36dab6895dfc14d026230592def51b78bb755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6facccc00e89c1176f35ee7a40af0743e8572a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6facccc00e89c1176f35ee7a40af0743e8572a20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
