)]}'
{"id":"openstack%2Fproject-config~870481","triplet_id":"openstack%2Fproject-config~master~Iaac5af66324cf780f07dca5886868f7cd540e4c6","project":"openstack/project-config","branch":"master","topic":"linaro-nodepool","hashtags":[],"change_id":"Iaac5af66324cf780f07dca5886868f7cd540e4c6","subject":"nb03/04: add linaro as a provider","status":"MERGED","created":"2023-01-16 01:56:18.000000000","updated":"2023-01-16 22:03:30.000000000","submitted":"2023-01-16 21:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870481-linaro-nodepool","meta_rev_id":"85494030685195a3049eb2b30700c1305d675929","_number":870481,"virtual_id_number":870481,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 21:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-16 21:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-16 21:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 02:07:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"683ebb9e3431ddde85173584c23284866ddf70a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 01:56:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83eff5a9e6da05ebff3e27ba96aa0354c8350c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:07:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4df86af575c4a29a6a8859e62dd3ead\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0e94dcf88a50400ba7924da85baa93d1 : SUCCESS in 2m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f110c1dc30242619ee334e606551927 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"626c2c17ef7d4bda2609e174cdc90eed0c530ce3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 21:09:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aada2232f0799f56b674f6b074500eff50b0c83b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 21:10:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d52e5640247decf40688873da5134e8a447887fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 21:10:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d4af75cbfd7d4192a89a3655aa503c6044831b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 21:23:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8a3423224144d5daac9e266d865ab93\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/40dfdee0c57643ac85f9dca1c58c7a47 : SUCCESS in 2m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/832c69f7767a495db78b06afa29f48fb : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7a9357c1bfe937e6782fcb93f2cd9981ff5be939","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 21:34:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ceb94d6cbd6119e4369eea65a51e6e6892dd479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 21:35:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ee1593966d01cb7acd9308cb18642ec23a46e05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 21:53:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b65a2241739d4fa5a7e3078d4fba79c2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3f069cd45ab04e59be1336e6de8474cd : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"8c5803457fc15cf003208279ea24a6b54f358aba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 21:53:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"85494030685195a3049eb2b30700c1305d675929","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 22:03:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79d5b17920b41e6ba030619f1666f99\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/cc7355b1662b4d408035c50dc01761cc : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac0375de48966c526e41645769d4c50f66b804af","revisions":{"a989d88eab5662a5b6270c9e4a92aa670834ef22":{"kind":"REWORK","_number":1,"created":"2023-01-16 01:56:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/81/870481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/870481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/870481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/870481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/870481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/870481/1"}}},"commit":{"parents":[{"commit":"1a855f3e9cd54988b1fb2199c85a98fb718fcffd","subject":"nl03 : reorganise labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a855f3e9cd54988b1fb2199c85a98fb718fcffd"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-16 01:54:46.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-16 01:54:46.000000000","tz":660},"subject":"nb03: add linaro as a provider","message":"nb03: add linaro as a provider\n\nFollow-on to I8c8439ca2ae8dd868bce9d2b015f6b8428b16a84 to add the new\nlinaro cloud as a provider.\n\nChange-Id: Iaac5af66324cf780f07dca5886868f7cd540e4c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a989d88eab5662a5b6270c9e4a92aa670834ef22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a989d88eab5662a5b6270c9e4a92aa670834ef22"}]},"branch":"refs/heads/master"},"ac0375de48966c526e41645769d4c50f66b804af":{"kind":"REWORK","_number":2,"created":"2023-01-16 21:10:55.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/81/870481/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/81/870481/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/870481/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/870481/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/81/870481/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/81/870481/2"}}},"commit":{"parents":[{"commit":"1a855f3e9cd54988b1fb2199c85a98fb718fcffd","subject":"nl03 : reorganise labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a855f3e9cd54988b1fb2199c85a98fb718fcffd"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-16 01:54:46.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-16 21:10:39.000000000","tz":660},"subject":"nb03/04: add linaro as a provider","message":"nb03/04: add linaro as a provider\n\nFollow-on to I8c8439ca2ae8dd868bce9d2b015f6b8428b16a84 to add the new\nlinaro cloud as a provider.\n\nChange-Id: Iaac5af66324cf780f07dca5886868f7cd540e4c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac0375de48966c526e41645769d4c50f66b804af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac0375de48966c526e41645769d4c50f66b804af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
