)]}'
{"id":"openstack%2Fproject-config~871006","triplet_id":"openstack%2Fproject-config~master~Ia30b3d45351c6cc97d51f32bc4820b72df5cfbb6","project":"openstack/project-config","branch":"master","topic":"linaro-nodepool","hashtags":[],"change_id":"Ia30b3d45351c6cc97d51f32bc4820b72df5cfbb6","subject":"nb04: use linaro region mirror","status":"MERGED","created":"2023-01-18 23:28:09.000000000","updated":"2023-01-19 00:48:02.000000000","submitted":"2023-01-19 00:39:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871006-linaro-nodepool","meta_rev_id":"52bee96b7a88250b7b9fb17da966cc63dec54afd","_number":871006,"virtual_id_number":871006,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 00:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-19 00:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-19 00:27:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 23:39:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"514aa30f0965c890d37723b1d30a1cbf6056545a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-18 23:28:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2109be58b1661aa3d5e1e1068c8aa1d2d2e32d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 23:39:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486f75083cd748e4bd73fa812184b4a7\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f2c44f2792db4c34a26824d75ec8d694 : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/94c624928d564f208cf2d5542e6cc1d5 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d7aca84ad21a7411d2521bf2356175bd11b141f8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-19 00:27:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14f2d75ab19dfe6d73791de83c366da581c76f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:28:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"750a400aa8afb06f763fe04e74582bfcd3db3668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:39:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8acc07fa1c14e64a6c36a85bb386fa1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/229eb8342c5e419a93e8eaf2cf5891f6 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"407200ed1cfdcc094784c42d5d7c5dcf462a67fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:39:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"52bee96b7a88250b7b9fb17da966cc63dec54afd","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:48:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3030f439a562431aae3967222feed01d\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/4e64fbb983b74402a6826ba6b063aa13 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af14c7e04be0f163e5ee1e72f4e77b085ea5682f","revisions":{"af14c7e04be0f163e5ee1e72f4e77b085ea5682f":{"kind":"REWORK","_number":1,"created":"2023-01-18 23:28:09.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/06/871006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/871006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/871006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/871006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/871006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/871006/1"}}},"commit":{"parents":[{"commit":"2f5f5db0b05913d662164b1038c1a861a11f0152","subject":"linaro-us : set max-servers to 0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f5f5db0b05913d662164b1038c1a861a11f0152"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-18 23:27:37.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-18 23:27:37.000000000","tz":660},"subject":"nb04: use linaro region mirror","message":"nb04: use linaro region mirror\n\nThis is pointing at the mirror in the old linaro region; point it at\nthe current one.\n\nChange-Id: Ia30b3d45351c6cc97d51f32bc4820b72df5cfbb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af14c7e04be0f163e5ee1e72f4e77b085ea5682f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af14c7e04be0f163e5ee1e72f4e77b085ea5682f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
