)]}'
{"id":"openstack%2Fproject-config~871196","triplet_id":"openstack%2Fproject-config~master~Ia049a2e44d2c4add0346e8262b60cdfb2c976539","project":"openstack/project-config","branch":"master","topic":"linaro-nodepool","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-01-23 21:15:36.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-01-24 22:00:57.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"Ia049a2e44d2c4add0346e8262b60cdfb2c976539","subject":"nodepool: drop linaro-us","status":"MERGED","created":"2023-01-19 21:55:17.000000000","updated":"2023-01-24 22:43:45.000000000","submitted":"2023-01-24 22:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871196-linaro-nodepool","meta_rev_id":"1995e26612b95c7c17576d03088321a1cd4dad7c","_number":871196,"virtual_id_number":871196,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 22:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-01-23 21:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-01-24 19:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-24 22:00:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 21:57:10.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2023-01-19 22:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 21:15:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-01-24 19:36:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"eebe6b89f29242b326f4329d403d7aabf13e1e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-19 21:55:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e04a144e8e88dc716239f29fb73e58ee4355ad6d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-19 21:57:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d1ca782ea075afd3858203d96c41ab8e71a5f679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 22:05:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49b321f19d9f477ba5b0cae7820cef9b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b8b1a00872014c69995cb47ba5b72bb2 : SUCCESS in 2m 40s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/911f28e163714f6cb28057d2800d3092 : SUCCESS in 6m 52s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/cb852fe17aef456a9c94947041dff7dd : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0309de66746e654dcbce0624be98aac657e6d955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-20 06:17:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90a495bb65c52da2be0f8bbcaebf1e72e1340b28","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-20 06:17:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5cc64949c464867ce85069fe8828b49b84ff64d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 06:31:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7db61f2975554343a57237d3ce35065d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b8955a03c4d9401386c45a932a74ca05 : SUCCESS in 2m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/08b0a0e762c84675b3556824c7412945 : SUCCESS in 6m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bdbfae6f01674c69b5af366a343de20c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"36e1d922389b0d91dcf4b30d902ef61cda490150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-23 04:48:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52037f0eddd712122e7d8c8f7c948dbac0e064a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 05:00:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c80a04fc721742999d82e7bcc97057b5\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/28b31b4af50e4171b283b4ff1684aed2 : SUCCESS in 2m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b65df23041464b8fa3d7b8c1252e89a1 : SUCCESS in 6m 35s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/171575a807cd4fedbc0f4c6936ae35a3 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6b7da202b7fdf714dcefc6bb9b106d6a54931aa2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-23 21:15:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bc6d6b658c40741176cb60c506ef89d5c23df3e0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-24 19:36:56.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c16fc75c99bda283adf708a36678d9688cfb5e05","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-24 22:00:57.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"91413796e0dea32aff3ce013e08848db63911f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:01:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bda8bda279dc2e76f512a3be355ae38d4df2243f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-24 22:01:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6142f9448566f524255003cd19099d0fd236afdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:06:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d414120c1f14f4fa12f542e40182294\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d153e3463e6442a68770c34d2d89a06b : SUCCESS in 5m 08s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e17da2b572b343eca64c292ed25ae865 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"31808d808b6c7615b1d895fdd68222a49731391d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:06:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1995e26612b95c7c17576d03088321a1cd4dad7c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:43:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9cf402924643c09fc2a791cc7b0f2a\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/ddf1dd950372402cb060836460d0f609 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"212ad67630ddc5e1e81217765d1ac72f41838f39","revisions":{"7b8cf4c37f19be510329d94e3dccce3ec84ca76b":{"kind":"REWORK","_number":1,"created":"2023-01-19 21:55:17.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/871196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/871196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/871196/1"}}},"commit":{"parents":[{"commit":"af14c7e04be0f163e5ee1e72f4e77b085ea5682f","subject":"nb04: use linaro region mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af14c7e04be0f163e5ee1e72f4e77b085ea5682f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-19 21:49:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-19 21:51:23.000000000","tz":660},"subject":"nodepool: drop linaro-us","message":"nodepool: drop linaro-us\n\nDrop the linaro-us cloud from nb04 uploads and launcher; it is\nreplaced by the new linaro cloud.  Region is not running any nodes\nsince I6ef17bb517078de363286ecad9749cb321b4c92c.\n\nnb03 is still in the inventory, but shutdown and in emergency.  We can\nremove the config here and cleanup will follow.\n\nChange-Id: Ia049a2e44d2c4add0346e8262b60cdfb2c976539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b8cf4c37f19be510329d94e3dccce3ec84ca76b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b8cf4c37f19be510329d94e3dccce3ec84ca76b"}]},"branch":"refs/heads/master"},"18cc824eac936db300d70ec102897e44a1d1a8c3":{"kind":"REWORK","_number":2,"created":"2023-01-20 06:17:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/871196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/871196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/871196/2"}}},"commit":{"parents":[{"commit":"d9a35d881bc871d153062cb6813ed667c85b4038","subject":"nodepool: empty linaro-us cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9a35d881bc871d153062cb6813ed667c85b4038"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-19 21:49:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-20 06:16:51.000000000","tz":660},"subject":"nodepool: drop linaro-us","message":"nodepool: drop linaro-us\n\nDrop the linaro-us cloud from nb04 uploads and launcher; it is\nreplaced by the new linaro cloud.  Region is not running any nodes\nsince I6ef17bb517078de363286ecad9749cb321b4c92c.\n\nnb03 is still in the inventory, but shutdown and in emergency.  We can\nremove the config here and cleanup will follow.\n\nChange-Id: Ia049a2e44d2c4add0346e8262b60cdfb2c976539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18cc824eac936db300d70ec102897e44a1d1a8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18cc824eac936db300d70ec102897e44a1d1a8c3"}]},"branch":"refs/heads/master"},"212ad67630ddc5e1e81217765d1ac72f41838f39":{"kind":"REWORK","_number":3,"created":"2023-01-23 04:48:44.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/871196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/871196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/871196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/871196/3"}}},"commit":{"parents":[{"commit":"0cf9319b0682d1f2228ef6bf8697f01c8056255c","subject":"nodepool: empty linaro-us cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0cf9319b0682d1f2228ef6bf8697f01c8056255c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-19 21:49:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-23 04:45:19.000000000","tz":660},"subject":"nodepool: drop linaro-us","message":"nodepool: drop linaro-us\n\nDrop the linaro-us cloud from nb04 uploads and launcher; it is\nreplaced by the new linaro cloud.  Region is not running any nodes\nsince I6ef17bb517078de363286ecad9749cb321b4c92c.\n\nnb03 is still in the inventory, but shutdown and in emergency.  We can\nremove the config here and cleanup will follow.\n\nChange-Id: Ia049a2e44d2c4add0346e8262b60cdfb2c976539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/212ad67630ddc5e1e81217765d1ac72f41838f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/212ad67630ddc5e1e81217765d1ac72f41838f39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
