)]}'
{"id":"openstack%2Fproject-config~871220","triplet_id":"openstack%2Fproject-config~master~Icf7b00f88a9de8a91510ee231c47eef207da4ea8","project":"openstack/project-config","branch":"master","topic":"linaro-nodepool","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-01-23 21:14:48.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-01-24 19:44:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf7b00f88a9de8a91510ee231c47eef207da4ea8","subject":"nodepool: empty linaro-us cloud","status":"MERGED","created":"2023-01-20 06:17:08.000000000","updated":"2023-01-24 19:52:42.000000000","submitted":"2023-01-24 19:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871220-linaro-nodepool","meta_rev_id":"417daec9ae51f5ff23c200343397c8824145ae6e","_number":871220,"virtual_id_number":871220,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 19:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-01-23 21:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-01-24 19:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2023-01-24 19:32:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-20 06:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-20 17:36:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-01-24 19:32:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"40ee8c32dd23f899786ed2c438f82e34ac5accc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-20 06:17:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0d8932cbdee2b7cf1ad112fd48f685478faa00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 06:30:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57c37f2cd97c40999742cff3f9604794\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e55fd3dac9cc41828754a2624ae6b1c6 : SUCCESS in 2m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2fe06ddbacbd4c70b79b8019c91b1ecc : SUCCESS in 6m 32s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1caf5558ce494a9b8bb95a31ab3fdf3e : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1fb88c395690e4a64d2cf3bb98901e98a17da6d3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-20 17:36:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2233e0e829f76bdcefc43563ba6655cd2dccce1c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-20 17:39:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fa07a8b883525ebab5cc5097c568cc5ef4469a7b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-23 04:34:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41975f2b53e7900bb841bd80575663a10b4fe080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-23 04:48:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21c45bca56706a1b0386a60742c7ea2070899962","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-23 04:53:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"765c4de8fac86df39c04765d099802d0466ca408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 05:00:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94454c496e0945cc98eacf95265097d5\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/79d3b13a9df048899c9f0b8568e0bdee : SUCCESS in 2m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/74c0db9de50b408e867199c2c2ad28d5 : SUCCESS in 6m 35s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ada38943295b46bbb1e6f4e0303fd2b7 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"419c11e966962074c7c38af0352bd010b9272146","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-23 21:14:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f96b3c49074e35a6270e074830e7a19859b5d6b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-01-24 19:32:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc3ed8c01ee477cb990a4872d9d196520c96c5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:33:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8a30abc63248170867877b72946460aef0a97ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:44:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d451e548555c4f8bb2a08117720fe465\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8d355b37df1e40e183e55d1eed771904 : SUCCESS in 6m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b77a1e67f0464be2a66a3e74734fc719 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f9c384da50440e71ba7cfe8be57613d68e083c50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:44:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"417daec9ae51f5ff23c200343397c8824145ae6e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:52:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ffb1600b36e4c599c128767a6416a4c\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/32b7276fc78f4034a489d8aba19dedad : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cf9319b0682d1f2228ef6bf8697f01c8056255c","revisions":{"d9a35d881bc871d153062cb6813ed667c85b4038":{"kind":"REWORK","_number":1,"created":"2023-01-20 06:17:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/20/871220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/871220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/871220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/871220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/871220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/871220/1"}}},"commit":{"parents":[{"commit":"af14c7e04be0f163e5ee1e72f4e77b085ea5682f","subject":"nb04: use linaro region mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af14c7e04be0f163e5ee1e72f4e77b085ea5682f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-20 06:13:38.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-20 06:16:28.000000000","tz":660},"subject":"nodepool: empty linaro-us cloud","message":"nodepool: empty linaro-us cloud\n\nThis should remove nodepool\u0027s tracking of the diskimages in this\ncloud, in preparation for it\u0027s removal.\n\nChange-Id: Icf7b00f88a9de8a91510ee231c47eef207da4ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9a35d881bc871d153062cb6813ed667c85b4038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9a35d881bc871d153062cb6813ed667c85b4038"}]},"branch":"refs/heads/master"},"0cf9319b0682d1f2228ef6bf8697f01c8056255c":{"kind":"REWORK","_number":2,"created":"2023-01-23 04:48:44.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/20/871220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/871220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/871220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/871220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/871220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/871220/2"}}},"commit":{"parents":[{"commit":"af14c7e04be0f163e5ee1e72f4e77b085ea5682f","subject":"nb04: use linaro region mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/af14c7e04be0f163e5ee1e72f4e77b085ea5682f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-20 06:13:38.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-23 04:45:00.000000000","tz":660},"subject":"nodepool: empty linaro-us cloud","message":"nodepool: empty linaro-us cloud\n\nThis should remove nodepool\u0027s tracking of the diskimages in this\ncloud, in preparation for it\u0027s removal.\n\nChange-Id: Icf7b00f88a9de8a91510ee231c47eef207da4ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0cf9319b0682d1f2228ef6bf8697f01c8056255c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0cf9319b0682d1f2228ef6bf8697f01c8056255c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
