)]}'
{"id":"openstack%2Fproject-config~871666","triplet_id":"openstack%2Fproject-config~master~I23328cbc53b87e1e81d26cc56f99aaad33b415c0","project":"openstack/project-config","branch":"master","topic":"linaro-nodepool","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-01-24 23:32:21.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I23328cbc53b87e1e81d26cc56f99aaad33b415c0","subject":"nodepool: fix new linaro provider name in nb04","status":"MERGED","created":"2023-01-24 22:07:41.000000000","updated":"2023-01-24 23:47:34.000000000","submitted":"2023-01-24 23:38:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871666-linaro-nodepool","meta_rev_id":"7a01e891365f6ce9a6be6dcb161c2127b2ae542f","_number":871666,"virtual_id_number":871666,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 23:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-01-24 22:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-24 23:32:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-24 22:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-24 22:20:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"1abc18d53984e568d4f972014af54b0fe10b4bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-24 22:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ed489f67ad59217a918c6b9ac44797810845d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 22:20:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3184eb0578a5464eac31231ee6d9bde2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9502798a86324f3798d89e670d2305d9 : SUCCESS in 2m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9c8a3b787054c0494bfee48a5cfe030 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"dfb73503b3c08eb6d5d436e7ab28a6ed2639e949","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-24 22:20:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27d135ae1ced99d04ab0a19e819bd2ecedc2cfd6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-24 23:32:21.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aae8265c2bc40c66066ab1a7d0b07af6a92c9f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 23:32:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5860f31e3a8072420a931a61a21486f0c1b107ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 23:38:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1e4e0747f445fa825ae61b97ad3b21\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/97a74194b07541968f949decd71b824b : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0578767fa9799ebcfd338f94fc69f5f2b6feabe7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 23:38:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a01e891365f6ce9a6be6dcb161c2127b2ae542f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 23:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc00b12771e4fc1a1b179a686f20ec0\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/109d51178b8642868b3844695b3e07b4 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd53b1c2b4324783d2b767784b7d6115b4b313ef","revisions":{"bd53b1c2b4324783d2b767784b7d6115b4b313ef":{"kind":"REWORK","_number":1,"created":"2023-01-24 22:07:41.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/66/871666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/871666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/871666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/871666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/871666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/871666/1"}}},"commit":{"parents":[{"commit":"212ad67630ddc5e1e81217765d1ac72f41838f39","subject":"nodepool: drop linaro-us","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/212ad67630ddc5e1e81217765d1ac72f41838f39"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-24 22:05:28.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-24 22:05:28.000000000","tz":660},"subject":"nodepool: fix new linaro provider name in nb04","message":"nodepool: fix new linaro provider name in nb04\n\nI have left the -regionone off this, making its naming inconsistent.\nThis adds it.\n\nSince this cloud is in its bringup phase, I will put the builder in\nemergency, clear out the images for the \"linaro\" provider and then\napply this by hand, so that we don\u0027t have old ZK nodes lying around.\nWe can then merge this to make it consistent.\n\nChange-Id: I23328cbc53b87e1e81d26cc56f99aaad33b415c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd53b1c2b4324783d2b767784b7d6115b4b313ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd53b1c2b4324783d2b767784b7d6115b4b313ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
