)]}'
{"id":"openstack%2Fproject-config~871703","triplet_id":"openstack%2Fproject-config~master~Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1","project":"openstack/project-config","branch":"master","topic":"retire-xstatic-font-awesome","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-02-03 16:21:58.000000000","reason":"Change was submitted"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-02-03 16:14:18.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2023-02-03 16:21:58.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-02-03 16:21:58.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-01 15:25:44.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1","subject":"Step 1: End Project Gating, Retire xstatic-font-awesome","status":"MERGED","created":"2023-01-25 11:03:53.000000000","updated":"2023-02-03 16:35:45.000000000","submitted":"2023-02-03 16:21:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"871703-retire-xstatic-font-awesome","meta_rev_id":"6909118463144e217a9e952cf6d80e8405501110","_number":871703,"virtual_id_number":871703,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-03 16:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2023-02-03 16:14:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-01 15:25:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-02-01 13:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2023-02-03 16:14:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 11:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-25 11:19:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 15:24:05.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-01-30 22:54:42.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-02-01 13:09:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-02-02 07:25:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-02-03 16:10:29.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"87922b7552d059c5a77c409d63f58fcfc7c71634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-25 11:03:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f2bc82dec54ebe9686b71990cc015bac2472252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 11:04:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job manila-tempest-dsvm-hdfs not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: x/devstack-plugin-hdfs\n      check:\n        jobs:\n          - openstack-tox-bashate\n          - legacy-tempest-dsvm-full-hdfs:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - manila-tempest-dsvm-hdfs:\n              voting: false\n              branches: ^(?!driverfixes/).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - openstack-tox-bashate\n          - legacy-tempest-dsvm-full-hdfs:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5839, column 3","accounts_in_message":[],"_revision_number":1},{"id":"ebc9f43587555544a847f434d7a6681bf715479f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-25 11:19:13.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"733d0475874ffb3bc1962d2deef841f4da13de8e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-01-25 15:24:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b013100e68f1a1522099d277a26b2674fa4b95a9","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-25 16:12:08.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b9a5829daff9ff3dd2f993cb20756c44971620b1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-30 22:54:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44ac87b68c6684e66f78bb8ca64519d1b0be4491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-31 12:48:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"857d94e5f0f14ee7e54ca1f58efdaed77e565714","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-31 12:49:01.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"b077fbd8765c05f49cc33950201829c61d3f9be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:49:07.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job manila-tempest-dsvm-hdfs not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: x/devstack-plugin-hdfs\n      check:\n        jobs:\n          - openstack-tox-bashate\n          - legacy-tempest-dsvm-full-hdfs:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - manila-tempest-dsvm-hdfs:\n              voting: false\n              branches: ^(?!driverfixes/).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - openstack-tox-bashate\n          - legacy-tempest-dsvm-full-hdfs:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5839, column 3","accounts_in_message":[],"_revision_number":2},{"id":"2d5f16668497cfb831dffbd4847135a569df5f3d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-31 12:51:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f00dff0bc0f7083f9d7858ecf9b39c2d95635926","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-01 11:47:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1459dd4ee0902f1b831fe18574721f2842c73c3f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 11:47:47.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":3},{"id":"8f3dcd89b7c4fd65f975728b3d88298509d403f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-01 11:49:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9107b0056c7e87aa38cd64b4cb8b8bfa886a576a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 11:49:18.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":4},{"id":"e4ee7c431785b6e8c50c499a458be8a13aa5ec21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 12:01:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4417336964a4baea43267aa680c68a7\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5a77fa3c94c747bea06e4417d2ee2786 : SUCCESS in 3m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c0c22322e6143c9a4b2e4dc7656fe61 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":4},{"id":"3514474e57ac1ec1c9cc14230f6466fa3b084149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-01 13:02:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6f81fdfc1ef6f35fe2ca9680d65b4c1c077feb50","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 13:02:29.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":5},{"id":"0a7565130c78aa84830622b1314e7f03ab768776","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-01 13:06:24.000000000","message":"Topic 871703 removed","accounts_in_message":[],"_revision_number":5},{"id":"c3fd7ba15a7430d307420216e48a85c98c5cbaf2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-01 13:06:29.000000000","message":"Topic set to retire-xstatic-font-awesome","accounts_in_message":[],"_revision_number":5},{"id":"c2de2863a55308442cdcfb572578c0a847013018","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-01 13:09:44.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9d35308e56646959ed129007191514fc58eb31f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 13:28:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d94ed374bfe34aaa8dd35cbd84d40d57\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8c92ede866034f88a39239f531c2611f : SUCCESS in 3m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6c322e08f8e04c8db4b64ecc7ae3a0a8 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":5},{"id":"48ad8100944cb7e93b754da8f94658845d706a03","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-01 15:25:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"08da651d96326235dbadcaac21c7f053ac9d4f78","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-03 16:14:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6c012f00717e8790bad9e1dc26b1abd48585550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:15:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3c6b6f6bd7245b63bf6f92a154fe5c0b55eb393c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:21:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ebb1dce56ed4b3a8dfacc66b7988394\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f235c266175b48bda1c06a4009cc569a : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":5},{"id":"7740931c01de57ef27228bd0ce4e3129139bb7df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:21:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6909118463144e217a9e952cf6d80e8405501110","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-03 16:35:45.000000000","message":"Patch Set 5:\n\nAlso the reason it\u0027s important not to remove the GitHub replication job at this stage is that we want the subsequent project change which replaces all repository content with a README about the retirement to propagate to the mirror on GitHub.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4c70de075c72f2c8c432f9c0b0b8fd8091d9e8a3","revisions":{"144003363a7d313e9297043bdf37a0f5a51db9e6":{"kind":"REWORK","_number":1,"created":"2023-01-25 11:03:53.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/03/871703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/871703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/871703/1"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-25 10:16:59.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-25 11:03:47.000000000","tz":330},"subject":"Retire xstatic-font-awesome and xstatic-jquery.tablesorter","message":"Retire xstatic-font-awesome and xstatic-jquery.tablesorter\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome and\nxstatic-jquery.tablesorter project from the OpenStack umbrella and\ncontributors from MoinMoin going to maintain this project under\npython-xstatic[3] or python-xstatic[4].\nThis patch is step 1 i.e. \"End Project Gating\" for retiring  process\nas mentioned in retiring a repository docs [5].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://github.com/python-xstatic\n[5] https://docs.openstack.org/project-team-guide/repository.html#step-1-end-project-gating\n\nChange-Id: Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/144003363a7d313e9297043bdf37a0f5a51db9e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/144003363a7d313e9297043bdf37a0f5a51db9e6"}]},"branch":"refs/heads/master"},"b92543fcebbc3956669e8f4d898f57dad27f2588":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-31 12:48:43.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/03/871703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/871703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/871703/2"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-25 10:16:59.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-31 12:43:16.000000000","tz":330},"subject":"Step 1: End Project Gating, Retire xstatic-font-awesome","message":"Step 1: End Project Gating, Retire xstatic-font-awesome\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome project from\nthe OpenStack umbrella and contributors from MoinMoin going to maintain\nthis project under python-xstatic[3].\n\nThis patch is step 1 i.e. \"End Project Gating\" for retiring  process\nas mentioned in retiring a repository docs [4].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://docs.openstack.org/project-team-guide/repository.html#step-1-end-project-gating\n\nChange-Id: Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b92543fcebbc3956669e8f4d898f57dad27f2588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b92543fcebbc3956669e8f4d898f57dad27f2588"}]},"branch":"refs/heads/master"},"d62432491e134081f6061bfc327c64c587496a0d":{"kind":"REWORK","_number":3,"created":"2023-02-01 11:47:29.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/03/871703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/871703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/871703/3"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-25 10:16:59.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-01 11:46:56.000000000","tz":330},"subject":"Step 1: End Project Gating, Retire xstatic-font-awesome","message":"Step 1: End Project Gating, Retire xstatic-font-awesome\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome project from\nthe OpenStack umbrella and contributors from MoinMoin going to maintain\nthis project under python-xstatic[3].\n\nThis patch is step 1 i.e. \"End Project Gating\" for retiring  process\nas mentioned in retiring a repository docs [4].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://docs.openstack.org/project-team-guide/repository.html#step-1-end-project-gating\n\nChange-Id: Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d62432491e134081f6061bfc327c64c587496a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d62432491e134081f6061bfc327c64c587496a0d"}]},"branch":"refs/heads/master"},"ff8ce11b8a93d6b5f12082b3404b9a2c3c33c138":{"kind":"REWORK","_number":4,"created":"2023-02-01 11:49:00.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/03/871703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/871703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/871703/4"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-25 10:16:59.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-01 11:48:31.000000000","tz":330},"subject":"Step 1: End Project Gating, Retire xstatic-font-awesome","message":"Step 1: End Project Gating, Retire xstatic-font-awesome\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome project from\nthe OpenStack umbrella and contributors from MoinMoin going to maintain\nthis project under python-xstatic[3].\n\nThis patch is step 1 i.e. \"End Project Gating\" for retiring  process\nas mentioned in retiring a repository docs [4].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://docs.openstack.org/project-team-guide/repository.html#step-1-end-project-gating\n\nChange-Id: Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff8ce11b8a93d6b5f12082b3404b9a2c3c33c138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ff8ce11b8a93d6b5f12082b3404b9a2c3c33c138"}]},"branch":"refs/heads/master"},"4c70de075c72f2c8c432f9c0b0b8fd8091d9e8a3":{"kind":"REWORK","_number":5,"created":"2023-02-01 13:02:11.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/03/871703/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/871703/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/871703/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/871703/5"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-25 10:16:59.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-01 13:01:19.000000000","tz":330},"subject":"Step 1: End Project Gating, Retire xstatic-font-awesome","message":"Step 1: End Project Gating, Retire xstatic-font-awesome\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome project from\nthe OpenStack umbrella and contributors from MoinMoin going to maintain\nthis project under python-xstatic[3].\n\nThis patch is step 1 i.e. \"End Project Gating\" for retiring  process\nas mentioned in retiring a repository docs [4].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://docs.openstack.org/project-team-guide/repository.html#step-1-end-project-gating\n\nChange-Id: Iba89d2fad269d0fba9042eb9864fb3ab31f6fbc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c70de075c72f2c8c432f9c0b0b8fd8091d9e8a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c70de075c72f2c8c432f9c0b0b8fd8091d9e8a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
