)]}'
{"id":"openstack%2Fproject-config~872222","triplet_id":"openstack%2Fproject-config~master~I8a88d6a9bcf5725986b00b063e03686d3225b48e","project":"openstack/project-config","branch":"master","topic":"add_fips_job","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-02-09 19:22:05.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2023-02-09 19:22:05.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-07 05:58:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-01-31 17:01:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I8a88d6a9bcf5725986b00b063e03686d3225b48e","subject":"Add base openstack FIPS job","status":"MERGED","created":"2023-01-30 21:36:02.000000000","updated":"2023-02-10 18:07:53.000000000","submitted":"2023-02-09 19:22:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872222-add_fips_job","meta_rev_id":"6a46b250b025a3af3c6165bbc9d2f3794827e8d3","_number":872222,"virtual_id_number":872222,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 19:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-07 05:58:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2023-02-07 13:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2023-02-09 19:12:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 21:36:25.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-30 21:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 21:44:44.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-01-30 21:44:44.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2023-01-30 21:44:44.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-02-10 15:23:00.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"af47dcb3391aad1b9fdce106a831f18729db8620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-30 21:36:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69ecb0ae2d7171057ee6dd46572b1c116b9a64ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-30 21:36:25.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"de5024d07eba8183d0517f1113c4aed8f381a916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:36:26.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job manila-tempest-dsvm-hdfs not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: x/devstack-plugin-hdfs\n      check:\n        jobs:\n          - openstack-tox-bashate\n          - legacy-tempest-dsvm-full-hdfs:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - manila-tempest-dsvm-hdfs:\n              voting: false\n              branches: ^(?!driverfixes/).*$\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      gate:\n        jobs:\n          - openstack-tox-bashate\n          - legacy-tempest-dsvm-full-hdfs:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5853, column 3","accounts_in_message":[],"_revision_number":1},{"id":"176c19457c3b95014168ad85f960d435b0117566","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-31 17:01:13.000000000","message":"Patch Set 1:\n\nrecheck now that the manila-tempest-dsvm-hdfs error is cleaned up","accounts_in_message":[],"_revision_number":1},{"id":"56494d3a2ea06c211fbb2ed2baef9ff3fe303b86","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-31 17:01:36.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"7b02252204990958e288d7c018b2315bd574627e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98a00aaa62bf44cea6a67e8aa58c33c7\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/afc316c6897948b4b7eb594cfc7ffe65 : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ab7814dc906640d98a199461705b8cec : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"93d2db42ee079da218864dad0a26b8337d151d1d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-31 18:46:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2152116b5014fc003ce50e64abea9c37094f5e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-31 22:20:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d476a7a64ebde59df8161834b162ab87b4b1e23b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-31 22:21:02.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"53a45678df59b8de4ccc194af05bed53130f95f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:24:33.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/866881 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"85850c7ba1de3cbe14acafa4848161532252071c","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-01-31 22:25:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2516fc88c1f1e782b333e4c1408a16812a3fb2f6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-31 22:25:35.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"24b3920daadca3ec6e4960498a60aef058d62582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:37:17.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13ecf6e2a7be4dadb68f197c04eb5bb8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/65c140a09433482cbdf64a484b9a7e18 : SUCCESS in 2m 59s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/728eee37eee04d9fb43068e56c117b94 : FAILURE in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"664fff65945cbd64f6af7640b29cc654638777ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-01 08:16:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2878732d32b3693c05f9c3e19ef8d1120ed5149c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 08:16:38.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":3},{"id":"6572f053defa0740b6048f237a743fc128cc4f20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-01 08:28:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4d93670496af3664794c67728b4fb37506ae417a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 08:29:08.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":4},{"id":"68e02b8619edaaeb4efe89e772e00a0ee7758bc9","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-01 08:29:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a374ea0c9c6d542c215f52124b09da2eda2dfd39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 08:29:55.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":4},{"id":"0d3e23d825af1dae63148e539cfee4786e8c5b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 08:46:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1bb9e65ce884b84861b8ed89feef2cf\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9e294be990f54f89b574703bdeddc9c9 : SUCCESS in 2m 56s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d95e4105ffc44ea9b0032cd8e8a55e17 : FAILURE in 7m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e26af9b9913c30b95696711110e9fdd8c71d8a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-01 08:59:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0f8121d56f60522bb7831ae004a7b5a9fbb39458","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-01 09:00:09.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":5},{"id":"42df91851ec005cb491de72c554d10fe31ec4afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 09:08:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/166bd0dd37ec43708e4935b96f8ed490\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/443979905c1f4d238400b691bbd60fc1 : SUCCESS in 3m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/22c64921038a4d04bf134140247a313f : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":5},{"id":"e35502a05685a1c3e47da3d27424a20bc3468eb7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-07 05:58:01.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2c97bc3f4f8a2a3192659ae4e6096fd50263203e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-07 13:48:23.000000000","message":"Patch Set 5: Code-Review+2\n\nNot approving since the Depends-On still needs another revision.","accounts_in_message":[],"_revision_number":5},{"id":"1521a49d8e1983085f54201e6ee81b90175e8916","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-09 19:12:55.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"388639a482d15bfd6d8e910dae0578b2ce9724c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 19:13:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"17a4292d81ff2ed390c500bbdafa415bfe3421e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 19:22:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/392636195ceb45c0b2351c74b57fb67a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/082414508938475a9cffcf995b1924d9 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":5},{"id":"c7bd0c2598d8c0098ac3e24e4bfbeb72030f9bd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 19:22:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"aedd61f1bb90b005497f93eb6623ea8a57169751","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-02-10 15:23:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b842a2cd16ce9b75111a5f279a7bbab5cf24eb5a","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2023-02-10 15:36:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf547f3221fe78aedfe5d0a5a8076ece7365d679","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-02-10 15:52:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a46b250b025a3af3c6165bbc9d2f3794827e8d3","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:07:53.000000000","message":"Patch Set 5:\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d3221813229b9366c50d48187253b9dc4eb839ca","revisions":{"fd62fa11d0c8fdbac5423473c0a7c2b34d89bbf6":{"kind":"REWORK","_number":1,"created":"2023-01-30 21:36:02.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/22/872222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/872222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/872222/1"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-27 16:10:39.000000000","tz":60},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-30 15:20:32.000000000","tz":60},"subject":"Add base openstack FIPS job","message":"Add base openstack FIPS job\n\nA new job openstack-fips is created.  This job is expected to be\na base job for most OpenStack jobs, so that FIPS testing can be\neasily enabled throughout OpenStack CI.\n\nA base job is required here because, for Ubuntu nodes, we need\nto enable an Ubuntu Advantage subscription, as Ubuntu considers\nFIPS to be a UA feature.  The subscription key is stored here in\nproject-config.\n\nDepends-On: I47a31f680172b47584510adb672b68498a85bd32\nChange-Id: I8a88d6a9bcf5725986b00b063e03686d3225b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd62fa11d0c8fdbac5423473c0a7c2b34d89bbf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd62fa11d0c8fdbac5423473c0a7c2b34d89bbf6"}]},"branch":"refs/heads/master"},"bbb4611b59b3f21c43dd194df4b4600c8972cb07":{"kind":"REWORK","_number":2,"created":"2023-01-31 22:20:35.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/22/872222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/872222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/872222/2"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-27 16:10:39.000000000","tz":60},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-31 22:19:56.000000000","tz":60},"subject":"Add base openstack FIPS job","message":"Add base openstack FIPS job\n\nA new job openstack-fips is created.  This job is expected to be\na base job for most OpenStack jobs, so that FIPS testing can be\neasily enabled throughout OpenStack CI.\n\nA base job is required here because, for Ubuntu nodes, we need\nto enable an Ubuntu Advantage subscription, as Ubuntu considers\nFIPS to be a UA feature.  The subscription key is stored here in\nproject-config.\n\nDepends-On: I47a31f680172b47584510adb672b68498a85bd32\nChange-Id: I8a88d6a9bcf5725986b00b063e03686d3225b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbb4611b59b3f21c43dd194df4b4600c8972cb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bbb4611b59b3f21c43dd194df4b4600c8972cb07"}]},"branch":"refs/heads/master"},"70aa6e37986f48194945a04d621e9072bf4fd8e0":{"kind":"REWORK","_number":3,"created":"2023-02-01 08:16:11.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/22/872222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/872222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/872222/3"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-27 16:10:39.000000000","tz":60},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-02-01 08:16:00.000000000","tz":60},"subject":"Add base openstack FIPS job","message":"Add base openstack FIPS job\n\nA new job openstack-fips is created.  This job is expected to be\na base job for most OpenStack jobs, so that FIPS testing can be\neasily enabled throughout OpenStack CI.\n\nA base job is required here because, for Ubuntu nodes, we need\nto enable an Ubuntu Advantage subscription, as Ubuntu considers\nFIPS to be a UA feature.  The subscription key is stored here in\nproject-config.\n\nDepends-On: I47a31f680172b47584510adb672b68498a85bd32\nChange-Id: I8a88d6a9bcf5725986b00b063e03686d3225b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70aa6e37986f48194945a04d621e9072bf4fd8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70aa6e37986f48194945a04d621e9072bf4fd8e0"}]},"branch":"refs/heads/master"},"173a1ded7e5063fe20bc7d9090a2eaad77e2a975":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-01 08:28:44.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/22/872222/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/872222/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/872222/4"}}},"commit":{"parents":[{"commit":"90c04fda837e3e0608b036dc385ce684c10e0947","subject":"Merge \"Reduce virtualpdu tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90c04fda837e3e0608b036dc385ce684c10e0947"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-27 16:10:39.000000000","tz":60},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-02-01 08:28:44.000000000","tz":0},"subject":"Add base openstack FIPS job","message":"Add base openstack FIPS job\n\nA new job openstack-fips is created.  This job is expected to be\na base job for most OpenStack jobs, so that FIPS testing can be\neasily enabled throughout OpenStack CI.\n\nA base job is required here because, for Ubuntu nodes, we need\nto enable an Ubuntu Advantage subscription, as Ubuntu considers\nFIPS to be a UA feature.  The subscription key is stored here in\nproject-config.\n\nDepends-On: I47a31f680172b47584510adb672b68498a85bd32\nChange-Id: I8a88d6a9bcf5725986b00b063e03686d3225b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/173a1ded7e5063fe20bc7d9090a2eaad77e2a975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/173a1ded7e5063fe20bc7d9090a2eaad77e2a975"}]},"branch":"refs/heads/master","description":"Rebase"},"d3221813229b9366c50d48187253b9dc4eb839ca":{"kind":"REWORK","_number":5,"created":"2023-02-01 08:59:45.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/22/872222/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/872222/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/872222/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/872222/5"}}},"commit":{"parents":[{"commit":"b913509ef4301ccf75ca3181f741f31deffec1e0","subject":"Merge \"nodepool: fix new linaro provider name in nb04\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b913509ef4301ccf75ca3181f741f31deffec1e0"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-01-27 16:10:39.000000000","tz":60},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2023-02-01 08:59:36.000000000","tz":60},"subject":"Add base openstack FIPS job","message":"Add base openstack FIPS job\n\nA new job openstack-fips is created.  This job is expected to be\na base job for most OpenStack jobs, so that FIPS testing can be\neasily enabled throughout OpenStack CI.\n\nA base job is required here because, for Ubuntu nodes, we need\nto enable an Ubuntu Advantage subscription, as Ubuntu considers\nFIPS to be a UA feature.  The subscription key is stored here in\nproject-config.\n\nDepends-On: I47a31f680172b47584510adb672b68498a85bd32\nChange-Id: I8a88d6a9bcf5725986b00b063e03686d3225b48e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3221813229b9366c50d48187253b9dc4eb839ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3221813229b9366c50d48187253b9dc4eb839ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
