)]}'
{"id":"openstack%2Fproject-config~872473","triplet_id":"openstack%2Fproject-config~master~I4e930080f89fe833702f7cafef09642e0638960f","project":"openstack/project-config","branch":"master","topic":"package-needs-drop-curl","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-02 22:00:21.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I4e930080f89fe833702f7cafef09642e0638960f","subject":"nodepool: infra-package-needs; drop curl","status":"MERGED","created":"2023-02-01 23:16:03.000000000","updated":"2023-02-02 22:46:49.000000000","submitted":"2023-02-02 22:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872473-package-needs-drop-curl","meta_rev_id":"5ce87261c744a1eecf54e48bb2ff3f22cf108172","_number":872473,"virtual_id_number":872473,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 22:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-02 20:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-02 22:00:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 23:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 20:46:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"0e56087df4592661a3cb0f31e943fc679437060a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-01 23:16:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"372457d41d7eb02d12a5c6a6acb293469f5af8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:27:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f13a836b70b4a8182c8a5bdbd3d938d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/80fa73357e1245e89c4e3df617b2b066 : SUCCESS in 2m 49s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0a10ce0fab9d47898fd6545d167fbb01 : SUCCESS in 6m 20s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/c0595fb1d73b48f693b9084ae4188a14 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"32ca0fd7b0a7641874f4e15ed50f418a762da0df","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-02 20:46:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65ae47402b65a36f0940a81668f02b4231e796d8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-02 22:00:21.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e931a8666ca959045ab3e43a2a9b3b217a6db4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 22:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"835686d1492768a2e1240278e9ff124dc8a03360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 22:08:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78032249a4fa4656a14b872e920a75b8\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26fe2a302f5341adb88d7593390f1be8 : SUCCESS in 6m 24s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/93843add08be44299303587be1b00f5b : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b67db37a06ab2b97a3db28559fbd5c731e3179db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 22:08:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ce87261c744a1eecf54e48bb2ff3f22cf108172","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 22:46:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6139e493b4ce45e1867766cea42cbdd8\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/405b9f76a29c4749bf92332695d26b69 : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90fcb99cf6f7cfb7a66c2124f17e8af5a0169159","revisions":{"90fcb99cf6f7cfb7a66c2124f17e8af5a0169159":{"kind":"REWORK","_number":1,"created":"2023-02-01 23:16:03.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/73/872473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/872473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/872473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/872473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/872473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/872473/1"}}},"commit":{"parents":[{"commit":"0083fc9cc227e7737b02f6faeda8b2eb86d96c22","subject":"Merge \"Add github sqlalchemy/sqlalchemy to the project list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0083fc9cc227e7737b02f6faeda8b2eb86d96c22"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:09:06.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:15:25.000000000","tz":660},"subject":"nodepool: infra-package-needs; drop curl","message":"nodepool: infra-package-needs; drop curl\n\nc.f. I9ccebe2dbf3a8682dab60c2070c5f78849e01446\n\nThe RedHat platforms vary if they come pre-installed with curl or\ncurl-minimal, and if curl-minimal is installed, it causes conflicts\nwhen you try to install \"curl\" (without removing it first, or using\n\"swap\").\n\npkg-map is not designed to deal with this at all; it can\u0027t say \"curl |\ncurl-minimal\".  But all our base images come with curl, because we\u0027re\nusing cache-url which uses it.\n\nSo, in short, drop it here to avoid this conflict.\n\nChange-Id: I4e930080f89fe833702f7cafef09642e0638960f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90fcb99cf6f7cfb7a66c2124f17e8af5a0169159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90fcb99cf6f7cfb7a66c2124f17e8af5a0169159"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
