)]}'
{"id":"openstack%2Fproject-config~872476","triplet_id":"openstack%2Fproject-config~master~I36942435a709c25097cb57d336c45c2884a0103c","project":"openstack/project-config","branch":"master","topic":"package-needs-drop-curl","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-02-02 21:13:05.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-06 00:06:00.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I36942435a709c25097cb57d336c45c2884a0103c","subject":"nodepool: infra-package-needs; cleanup python","status":"MERGED","created":"2023-02-01 23:58:32.000000000","updated":"2023-02-06 00:42:14.000000000","submitted":"2023-02-06 00:16:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872476-package-needs-drop-curl","meta_rev_id":"d099e0959ee746e1ef0a23204ad56a8626dea0f1","_number":872476,"virtual_id_number":872476,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-06 00:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-02 21:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-06 00:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 00:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 20:43:48.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2023-02-02 21:13:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"b5eedafd2944c14c642eeaf4a16c1956f5c9de24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-01 23:58:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9e37a84b30805786347d3261af4e4fb0abc1eeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:17:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/012a2f305fe44a058ba04c0f1d5357cb\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2ee331f6c8fc468b9ab3351310d2c256 : SUCCESS in 3m 06s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/611f2e39d0d54d60bb18a17821fec8d1 : SUCCESS in 8m 44s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/7720a1fe2a4d4074b35c7d9b990c30b4 : FAILURE in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"28756e09fe9c2232f2eff6a9d5fe200c1ea4d452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-02 00:24:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d1a1e59ed0278bfee5538bce3a1f547917ce13d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:33:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65e7cf2aff6049f29edeff930381055e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/66a4dc3e160246a696b6b601b85f6dc2 : SUCCESS in 4m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b0b648acbdab4106811ae6cb28cc6847 : SUCCESS in 5m 58s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/1a22c5682c58428297a5205fc64385a6 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ebfed745fc014639fce9a004ed51544fdf3ab9e0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-02 20:43:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d23e494a2e1e1d237b0bcbb6fafed23a896d208","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-02 21:03:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19e043512ee57386d2f87797e8274083c979983a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-02 21:13:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d275536fbdb6fa50c040f500b062c2d4606d5121","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-06 00:06:00.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0291ac9f38b1624a05f21f2207dea175d6509a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:06:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5e2c8f107f50d3f56a7a55b56813143a682b2afc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:16:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da186d0eb6144ca183b85eef6bb53fd1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f104b45c9c534013a1315155cf4fa82a : SUCCESS in 5m 50s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/3dac4951d0264af3ae00ac2eec52d7a1 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b114fbf0702ec953b46a1bc62018022792c0d9ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:16:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d099e0959ee746e1ef0a23204ad56a8626dea0f1","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:42:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f52ad507134016a1f77e0eea1ae907\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/b11aefb70e9a4b2c802efe0737ef30b3 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4437dcd0fd879da6c982ff68f83dd57fc6984d81","revisions":{"1eb2f944fb45bc5e19f705b380c720ab4bf418c7":{"kind":"REWORK","_number":1,"created":"2023-02-01 23:58:32.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/76/872476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/872476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/872476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/872476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/872476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/872476/1"}}},"commit":{"parents":[{"commit":"90fcb99cf6f7cfb7a66c2124f17e8af5a0169159","subject":"nodepool: infra-package-needs; drop curl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90fcb99cf6f7cfb7a66c2124f17e8af5a0169159"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:24:28.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:31:37.000000000","tz":660},"subject":"nodepool: infra-package-needs; cleanup python","message":"nodepool: infra-package-needs; cleanup python\n\nWe don\u0027t need to pull in Python 2 python-xml or python-dev packages\nany more\n\npython3 is always installed by DIB (it needs python3 on the image to\nrun elements).  So we don\u0027t explicitly need to pull that in.\n\nChange-Id: I36942435a709c25097cb57d336c45c2884a0103c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1eb2f944fb45bc5e19f705b380c720ab4bf418c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1eb2f944fb45bc5e19f705b380c720ab4bf418c7"}]},"branch":"refs/heads/master"},"4437dcd0fd879da6c982ff68f83dd57fc6984d81":{"kind":"REWORK","_number":2,"created":"2023-02-02 00:24:39.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/76/872476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/872476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/872476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/872476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/872476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/872476/2"}}},"commit":{"parents":[{"commit":"90fcb99cf6f7cfb7a66c2124f17e8af5a0169159","subject":"nodepool: infra-package-needs; drop curl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90fcb99cf6f7cfb7a66c2124f17e8af5a0169159"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:24:28.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-02 00:24:21.000000000","tz":660},"subject":"nodepool: infra-package-needs; cleanup python","message":"nodepool: infra-package-needs; cleanup python\n\nWe don\u0027t need to pull in Python 2 python-xml or python-dev packages\nany more\n\npython3 is always installed by DIB (it needs python3 on the image to\nrun elements).  So we don\u0027t explicitly need to pull that in.\n\nChange-Id: I36942435a709c25097cb57d336c45c2884a0103c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4437dcd0fd879da6c982ff68f83dd57fc6984d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4437dcd0fd879da6c982ff68f83dd57fc6984d81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
