)]}'
{"id":"openstack%2Fproject-config~872477","triplet_id":"openstack%2Fproject-config~master~I0b8908a79a5dcbe2a5bf5bf72986ea28e17c95fa","project":"openstack/project-config","branch":"master","topic":"package-needs-drop-curl","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-06 00:16:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b8908a79a5dcbe2a5bf5bf72986ea28e17c95fa","subject":"nodepool: infra-package-needs; remove lvm2","status":"MERGED","created":"2023-02-01 23:58:32.000000000","updated":"2023-02-06 00:52:03.000000000","submitted":"2023-02-06 00:16:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872477-package-needs-drop-curl","meta_rev_id":"c513333b5c9c4d0ebcdde46126a9b36e9fcce805","_number":872477,"virtual_id_number":872477,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-06 00:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-02 20:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-02-03 10:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2023-02-03 10:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 00:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 20:44:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-02-03 10:57:52.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"ac562363cd968295b3f5cf9aa290997082a09195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-01 23:58:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a940b8d47d3d119d9508ef4a6fd41d5089833d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:04:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbc7072938804afa9f057898cc8860d3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b9b6ee0dcd5f42b7a4184affef7bef72 : SUCCESS in 3m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0dbd3dd575b84a388cce989403b104b5 : SUCCESS in 5m 05s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/8e87bf1f7e9f480ea255e622476264b1 : FAILURE in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a7b71b1bda28935f2eda9405b1257ff1d72dd0e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-02 00:24:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1169e88f4e11da6d3e20dad2b3e5c45227a027c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:42:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9fede22f2934ecfb78ab75dd76b5955\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6bbbaf5d845441b4a38a7dcbc3c82bcb : SUCCESS in 4m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca99f0ee7929492da874ed0606312aab : SUCCESS in 10m 50s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/41b218497075428ca42e579bc71b296a : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"452be00627652ae766f0d3d950b7356c2f8fa874","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-02 20:44:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c896d4aa10159e1756bfc188cdf26f5a68e7b2e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-02-03 10:57:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5623d0475ce45276907e2b0c871c431c9061f5ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:06:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a744a4f1986ace48a237243b375bf83714f97593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:16:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5a3107b6174441bc68bb528701003e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c9631b42f0ed43ffa665e8e443f69401 : SUCCESS in 6m 04s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/337261bbba91472ba9884548d377f611 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f96efd8f540bbb2a2028c591fe2a569286bf0d9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:16:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c513333b5c9c4d0ebcdde46126a9b36e9fcce805","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:52:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6716534a69d64120ac94045235ea755a\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/bb2655eccc7a44e0a60275b53557be1f : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a6b14875f4ed20d422ba208c5134f18d9b796da","revisions":{"c0d357b036324fdddb28aebdbd2588d3c0ec2dad":{"kind":"REWORK","_number":1,"created":"2023-02-01 23:58:32.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/77/872477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/872477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/872477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/872477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/872477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/872477/1"}}},"commit":{"parents":[{"commit":"1eb2f944fb45bc5e19f705b380c720ab4bf418c7","subject":"nodepool: infra-package-needs; cleanup python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1eb2f944fb45bc5e19f705b380c720ab4bf418c7"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:33:23.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:33:23.000000000","tz":660},"subject":"nodepool: infra-package-needs; remove lvm2","message":"nodepool: infra-package-needs; remove lvm2\n\nThis came in via Ie1a0aba57390c9c0b269b4cbb076090ae1de73a9 many years\nago, when it was copied from old puppet.  I can\u0027t see that we need to\nbe installing this for any infra reason.\n\nI guess there is a small posibility things are relying on this, but\nthey would be better to install it themselves anyway.\n\nChange-Id: I0b8908a79a5dcbe2a5bf5bf72986ea28e17c95fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0d357b036324fdddb28aebdbd2588d3c0ec2dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0d357b036324fdddb28aebdbd2588d3c0ec2dad"}]},"branch":"refs/heads/master"},"5a6b14875f4ed20d422ba208c5134f18d9b796da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-02 00:24:39.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/77/872477/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/872477/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/872477/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/872477/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/872477/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/872477/2"}}},"commit":{"parents":[{"commit":"4437dcd0fd879da6c982ff68f83dd57fc6984d81","subject":"nodepool: infra-package-needs; cleanup python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4437dcd0fd879da6c982ff68f83dd57fc6984d81"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:33:23.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-02 00:24:24.000000000","tz":660},"subject":"nodepool: infra-package-needs; remove lvm2","message":"nodepool: infra-package-needs; remove lvm2\n\nThis came in via Ie1a0aba57390c9c0b269b4cbb076090ae1de73a9 many years\nago, when it was copied from old puppet.  I can\u0027t see that we need to\nbe installing this for any infra reason.\n\nI guess there is a small posibility things are relying on this, but\nthey would be better to install it themselves anyway.\n\nChange-Id: I0b8908a79a5dcbe2a5bf5bf72986ea28e17c95fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a6b14875f4ed20d422ba208c5134f18d9b796da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a6b14875f4ed20d422ba208c5134f18d9b796da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[]}
