)]}'
{"id":"openstack%2Fproject-config~872478","triplet_id":"openstack%2Fproject-config~master~If3fddd79dde56f4087e465ed8b8013f0f337e0cb","project":"openstack/project-config","branch":"master","topic":"package-needs-drop-curl","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-07 06:19:15.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"If3fddd79dde56f4087e465ed8b8013f0f337e0cb","subject":"nodepool: infra-package-needs; cleanup tox installs","status":"MERGED","created":"2023-02-01 23:58:32.000000000","updated":"2023-02-07 06:48:55.000000000","submitted":"2023-02-07 06:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"872478-package-needs-drop-curl","meta_rev_id":"17a48ffc84a816f99866d9f326e124e0d9276084","_number":872478,"virtual_id_number":872478,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 06:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-02 20:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-07 06:19:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 00:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 20:57:20.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"30f7df4c84fb5e2e1caf9f06f44e8af5dcf3e236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-01 23:58:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6ea5ef27503974abaa43c58228b46bdbf76a262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:17:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e5f07d81c964c8bb91664f1e0ebfd24\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3b0fc81745934b13b80ff94b6148dabf : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e75ef3f461714eb4b254eb54cc495bbf : SUCCESS in 10m 25s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/824c1241654145c78e7394c0648e1975 : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ed86b379f04be660411379f1da43fead931bc4e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-02 00:24:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a7844d540f9b2f85c71c1d0a4a05bb965c5e7187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:36:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e97ac0e3f1045508e1db6d7121ca5a7\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ae32db95b30a4679b58e445705f8eef8 : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a31eb59668aa411db7a4fd7b8048bf5b : SUCCESS in 6m 46s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/93a79abded1440cda5c1f72d3a8f2e37 : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a58f4f4878249a43f1a0b1c551af0dac47ac7925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-02 00:46:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"028d157231e5e4a5f12f0e0346c99e6e73d19bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:55:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f53de07ac54d42c1823a6b61ffb802c0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/35901047f75e40ae9972fec85a18c7b3 : SUCCESS in 5m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/67689decd1184d799c32bcaa43d8f252 : SUCCESS in 6m 51s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/1c0c6e743aaf4d34b2dd7809f351941d : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b676fbc0556bf6ed9231a1fe9df3e2a777017691","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-02 20:57:20.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ae9c8d9d4b211a325a843d5e6ff557e7cae3a3bc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-07 06:19:15.000000000","message":"Patch Set 3: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d8ee5a633c634003b2abaf00b44e1ce38d822bae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 06:19:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a137f596177175b6d3342bc3e02ea257a084f46a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 06:26:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb5e694410e4c14834169e38a1f7088\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5954da6a1fd54fd3a87d1bcca9d75d1b : SUCCESS in 5m 46s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/3c817047c3d54e6c951578b043e0612f : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"d8567a9960c6f92106678241f842f1fc2935623b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 06:26:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"17a48ffc84a816f99866d9f326e124e0d9276084","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 06:48:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/771c6b03ceb0424eb99d1a7c2077f05b\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/f726f06885364c9e9c836bfb3e6aad8c : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"75a6a641b11a064f85e5620f9ad6bc35106139ca","revisions":{"0b5e31605d7970196f407a8a945742b309f261e6":{"kind":"REWORK","_number":1,"created":"2023-02-01 23:58:32.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/78/872478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/872478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/872478/1"}}},"commit":{"parents":[{"commit":"c0d357b036324fdddb28aebdbd2588d3c0ec2dad","subject":"nodepool: infra-package-needs; remove lvm2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c0d357b036324fdddb28aebdbd2588d3c0ec2dad"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:51:46.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:58:17.000000000","tz":660},"subject":"nodepool: infra-package-needs; cleanup tox installs","message":"nodepool: infra-package-needs; cleanup tox installs\n\nThe package-maps install of tox is only defined for gentoo, and that\ncame in with the original image build parts.  We don\u0027t need that any\nmore.\n\n10-pip-packages I didn\u0027t trace down, but it hasn\u0027t been doing anything\nfor a long time, since we removed pip-and-virtualenv.  We can remove\nthat.\n\nThe install done in 40-install-tox I can not see being used anywhere.\nIt came in with If5397d731e9fb04431482529aed23cd9fdaecc1d but I can\u0027t\nsee the venv actually referenced anywhere.  I think this has all been\nreplaced by the ensure-tox role (or, indeed, jobs migrating away from\ntox).  Remove it.\n\nChange-Id: If3fddd79dde56f4087e465ed8b8013f0f337e0cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b5e31605d7970196f407a8a945742b309f261e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b5e31605d7970196f407a8a945742b309f261e6"}]},"branch":"refs/heads/master"},"31ebc29d5757d9ab59ffb029a1ad864e7a5a4b80":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-02 00:24:39.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/78/872478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/872478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/872478/2"}}},"commit":{"parents":[{"commit":"5a6b14875f4ed20d422ba208c5134f18d9b796da","subject":"nodepool: infra-package-needs; remove lvm2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a6b14875f4ed20d422ba208c5134f18d9b796da"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:51:46.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-02 00:24:24.000000000","tz":660},"subject":"nodepool: infra-package-needs; cleanup tox installs","message":"nodepool: infra-package-needs; cleanup tox installs\n\nThe package-maps install of tox is only defined for gentoo, and that\ncame in with the original image build parts.  We don\u0027t need that any\nmore.\n\n10-pip-packages I didn\u0027t trace down, but it hasn\u0027t been doing anything\nfor a long time, since we removed pip-and-virtualenv.  We can remove\nthat.\n\nThe install done in 40-install-tox I can not see being used anywhere.\nIt came in with If5397d731e9fb04431482529aed23cd9fdaecc1d but I can\u0027t\nsee the venv actually referenced anywhere.  I think this has all been\nreplaced by the ensure-tox role (or, indeed, jobs migrating away from\ntox).  Remove it.\n\nChange-Id: If3fddd79dde56f4087e465ed8b8013f0f337e0cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/31ebc29d5757d9ab59ffb029a1ad864e7a5a4b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/31ebc29d5757d9ab59ffb029a1ad864e7a5a4b80"}]},"branch":"refs/heads/master"},"75a6a641b11a064f85e5620f9ad6bc35106139ca":{"kind":"REWORK","_number":3,"created":"2023-02-02 00:46:36.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/78/872478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/872478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/872478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/872478/3"}}},"commit":{"parents":[{"commit":"5a6b14875f4ed20d422ba208c5134f18d9b796da","subject":"nodepool: infra-package-needs; remove lvm2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a6b14875f4ed20d422ba208c5134f18d9b796da"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-01 23:51:46.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-02 00:46:16.000000000","tz":660},"subject":"nodepool: infra-package-needs; cleanup tox installs","message":"nodepool: infra-package-needs; cleanup tox installs\n\nThe package-maps install of tox is only defined for gentoo, and that\ncame in with the original image build parts.  We don\u0027t need that any\nmore.\n\n10-pip-packages I didn\u0027t trace down, but it hasn\u0027t been doing anything\nfor a long time, since we removed pip-and-virtualenv.  We can remove\nthat.\n\nThe install done in 40-install-tox I can not see being used anywhere.\nIt came in with If5397d731e9fb04431482529aed23cd9fdaecc1d but I can\u0027t\nsee the venv actually referenced anywhere.  I think this has all been\nreplaced by the ensure-tox role (or, indeed, jobs migrating away from\ntox).  Remove it.\n\nChange-Id: If3fddd79dde56f4087e465ed8b8013f0f337e0cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/75a6a641b11a064f85e5620f9ad6bc35106139ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/75a6a641b11a064f85e5620f9ad6bc35106139ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
