)]}'
{"id":"openstack%2Fproject-config~872835","triplet_id":"openstack%2Fproject-config~master~I056acc1b3847c48aaa76a35475da74b05ca24cb2","project":"openstack/project-config","branch":"master","topic":"retire-xstatic-font-awesome","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-02-17 20:43:12.000000000","reason":"Change was submitted"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-02-17 20:43:12.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-17 20:43:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I056acc1b3847c48aaa76a35475da74b05ca24cb2","subject":"Step 3: Remove project from infra System - Retire xstatic-font-awesome","status":"MERGED","created":"2023-02-07 03:58:26.000000000","updated":"2023-02-17 21:05:37.000000000","submitted":"2023-02-17 20:43:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872835-retire-xstatic-font-awesome","meta_rev_id":"c7949bebdb92c804756af3025138d493c2593651","_number":872835,"virtual_id_number":872835,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 20:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-17 21:03:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-02-15 22:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"date":"2023-02-17 20:29:47.000000000","_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2023-02-17 20:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-07 03:58:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-07 04:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 18:05:43.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-02-15 22:57:58.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-02-16 02:03:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"8a9c3af196c44ca638ed3786674320eaf5683275","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-07 03:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"959e4b61fc73e5eac7d3a361d135d4c1bb21f09b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-07 03:58:39.000000000","message":"Topic 871703 removed","accounts_in_message":[],"_revision_number":1},{"id":"3f220feb68518da53259fbe4332da42fb6bfa351","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-07 03:58:45.000000000","message":"Topic set to retire-xstatic-font-awesome","accounts_in_message":[],"_revision_number":1},{"id":"bde4f4b0afb71c1793adafd613575f3f5b3bc676","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-07 03:58:45.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"8b93a509ce7ddcfadf0afcd4fdf27bdc82b52c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 04:20:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e43ec8a3fe6a421596eb4e922496ecf8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/79b5fa0755544b0a90791168e92d94f9 : SUCCESS in 3m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c604fa45df804b8a9af9b0cf6e634b23 : SUCCESS in 8m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/718041cc318b4dcdb373942a9faff37e : SUCCESS in 3m 38s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c49bfa7ad1be45d68e50899b314a02b9 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9295a1229d564102e067017d5e3146e91f694c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 16:35:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bdb1b8d03af2cb6d8054b11610109df42edf975f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-15 16:35:35.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"04e4e4ba57f2af37e554d39dff89b2a2345808d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 16:50:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26d94bc03df549d8939868bb8b69d30f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/485388160a71402e9082ceac13e4caf7 : SUCCESS in 3m 04s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f0388ee2ad1e4662802ecb6c44c8a041 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9d9b92e007454183bf72fd76c08fa485 : SUCCESS in 3m 27s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/2cd85198e5c44161ab41202658f33b6a : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d590f30277e559bba7f09a78eeb91a571d9ef1c0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-15 22:57:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7bc23e2f656426003c5cd2f8e545c568c35ca185","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 02:03:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e83ae277db5f53f0b1e15125a10eb3472955fa23","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 18:51:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70f84250fb6eb854a5e7587f7320fcd91d3a7b79","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-16 19:23:31.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9b34063cc5bdbb63e4a3340725b7200291a3c4ee","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-16 19:23:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"07afbc56b21344c11a5bd210b6fa0678b3e0d7d1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-17 20:29:47.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5922c5b61d83fc4b477da4591cdb402459384f63","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-17 20:35:28.000000000","message":"Patch Set 2: Workflow+1\n\nInterpreting Ghanshyam\u0027s -1 as a +2 now that his objection has been addressed.","accounts_in_message":[],"_revision_number":2},{"id":"8ed488a157de79d3f5d5e6f017c11103922768b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:36:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac02867cea86f03d9bcd153eecb76b8f1dae4375","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:43:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f54e2cedf64343938a78a9c19d0fa1dd\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2549c127543543809243a3fed93d699e : SUCCESS in 5m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/32851234a2ae45d49af5a35952398e6a : SUCCESS in 3m 32s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/385d925aa47541be90e6ff18492218f4 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ed59ae25dad92d3bff7075bf3c40458edb211549","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 20:43:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"412faf540432948157f5c0e5a2aeeb26c8a21e69","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-17 21:03:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7949bebdb92c804756af3025138d493c2593651","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 21:05:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619556c3ed5b432bba3bd06cdb2e3832\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/af632e441fe84b709ec5aad7deb0c8a0 : SUCCESS in 5m 59s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/48ae5ffc9dfd4ff8b50ad037f3993419 : SUCCESS in 8m 23s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/154e0e7715f74ca18b1775929577dedc : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f5b11df0ccc126990bdb90aa4df6561fa967020","revisions":{"5c146adf1915395b813bf827af2e60b50c93d4a5":{"kind":"REWORK","_number":1,"created":"2023-02-07 03:58:26.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/35/872835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/872835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/872835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/872835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/872835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/872835/1"}}},"commit":{"parents":[{"commit":"f30f6c536fe3e6268cad8ce97dbc7535e213c185","subject":"Create a git for the storage of public keys and certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f30f6c536fe3e6268cad8ce97dbc7535e213c185"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-07 03:51:30.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-07 03:51:30.000000000","tz":330},"subject":"Retire xstatic-font-awesome: remove project from infra","message":"Retire xstatic-font-awesome: remove project from infra\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome project from\nthe OpenStack umbrella and contributors from MoinMoin going to maintain\nthis project under python-xstatic[3].\n\nThis patch is step 3 i.e. \"Remove Project from Infrastructure Systems\"\nfor retiring  process as mentioned in retiring a repository docs [4].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#step-3-remove-project-from-infrastructure-systems\n\nDepends-On: https://review.opendev.org/c/openstack/xstatic-font-awesome/+/872106\nChange-Id: I056acc1b3847c48aaa76a35475da74b05ca24cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c146adf1915395b813bf827af2e60b50c93d4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c146adf1915395b813bf827af2e60b50c93d4a5"}]},"branch":"refs/heads/master"},"6f5b11df0ccc126990bdb90aa4df6561fa967020":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-15 16:35:16.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/35/872835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/872835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/872835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/872835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/872835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/872835/2"}}},"commit":{"parents":[{"commit":"f30f6c536fe3e6268cad8ce97dbc7535e213c185","subject":"Create a git for the storage of public keys and certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f30f6c536fe3e6268cad8ce97dbc7535e213c185"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-07 03:51:30.000000000","tz":330},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2023-02-15 16:35:15.000000000","tz":0},"subject":"Step 3: Remove project from infra System - Retire xstatic-font-awesome","message":"Step 3: Remove project from infra System - Retire xstatic-font-awesome\n\nAs discussed in the horizon weekly meeting[1] and announced on open-discuss\nML [2], Horizon team decided to retire xstatic-font-awesome project from\nthe OpenStack umbrella and contributors from MoinMoin going to maintain\nthis project under python-xstatic[3].\n\nThis patch is step 3 i.e. \"Remove Project from Infrastructure Systems\"\nfor retiring  process as mentioned in retiring a repository docs [4].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-01-18.log.html#t2023-01-18T15:05:56\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2023-January/031884.html\n[3] https://github.com/python-xstatic/font-awesome\n[4] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#step-3-remove-project-from-infrastructure-systems\n\nDepends-On: https://review.opendev.org/c/openstack/xstatic-font-awesome/+/872106\nChange-Id: I056acc1b3847c48aaa76a35475da74b05ca24cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f5b11df0ccc126990bdb90aa4df6561fa967020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f5b11df0ccc126990bdb90aa4df6561fa967020"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
