)]}'
{"id":"openstack%2Fproject-config~873244","triplet_id":"openstack%2Fproject-config~master~Ic8b3e790fe332cf68bad7aaa3d5f85229600380b","project":"openstack/project-config","branch":"master","topic":"cache-cirros0.6.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-02-14 17:34:50.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-02-14 17:34:50.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-14 17:34:50.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-14 17:34:50.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-02-09 12:49:34.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic8b3e790fe332cf68bad7aaa3d5f85229600380b","subject":"Cache Cirros 0.6.1 images","status":"MERGED","created":"2023-02-09 11:36:40.000000000","updated":"2023-02-14 17:47:45.000000000","submitted":"2023-02-14 17:34:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873244-cache-cirros0.6.1","meta_rev_id":"17f0c8d9f98a9b494e829152bd209ec02f213150","_number":873244,"virtual_id_number":873244,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 17:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-14 17:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-09 12:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-02-14 17:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-09 11:46:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-09 11:59:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2023-02-09 11:59:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-02-09 11:59:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-02-14 14:21:44.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-02-14 17:24:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"9f4ef2329f384836378be8071847a9814f9a5f79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-09 11:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d18e71e81d53495f774fad3501907bb6291d8c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 11:46:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2dc91d1e6e4b71a682d63bb3aaf8e7\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/793216f4e4884025b73a53b2543fa876 : SUCCESS in 3m 07s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/588c11ca76654b658dc6be444d164ad7 : SUCCESS in 7m 04s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/02727609371d4924881b717aa0e6e86c : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a5f9d54d691fa8be523d6b8cd5021519fb6bf649","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-09 12:49:34.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is also a good opportunity to clean up older unused images from the cache so we can free up space and shrink our node images (can be done in a separate change of course). Do you happen to know how to go about figuring out which ones are no longer in use?","accounts_in_message":[],"_revision_number":1},{"id":"46e19bc7dc8e2f9f961ebdfcd46869f8bc76b66a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-09 14:54:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a489c4143fdbb81631afac843c4dfab7234d04d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-09 14:57:34.000000000","message":"Patch Set 1:\n\nThanks! Also remember codesearch doesn\u0027t search non-default branches in projects, so wouldn\u0027t pick up stable branch use of these. Probably we could query OpenSearch to see if jobs are actively booting those images, since we should only need to cache the ones which are getting used with any regularity. Something to think about, I suppose.","accounts_in_message":[],"_revision_number":1},{"id":"688eb2524e6f0f60bf836d002731098345973026","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-02-09 15:15:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c615eb8901f5e935244e4f25614fe4b72bb6a8f9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-14 17:24:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c52cfd21f18c16bb5b74a4492cffda0baad04d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 17:25:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c181b3b5916d1d91ad4a5e20c6b9f3051a02f57","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-14 17:28:37.000000000","message":"Patch Set 1:\n\nI\u0027ve proposed Ibada405e0c1183559f428c749d0e54d0a45a2223 to clean up the older images we\u0027re not using often.","accounts_in_message":[],"_revision_number":1},{"id":"ca49df23ce9f6c174e1d03a73a7683777a534a4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 17:34:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c0438976234035877f40de26a2084e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a324451aa6e443b69f322db5010f4544 : SUCCESS in 7m 50s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/33b9f8b2e86142e09524fc5fff7b9703 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f3f06d05f672253157eacde008185fdc033a6186","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 17:34:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"17f0c8d9f98a9b494e829152bd209ec02f213150","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 17:47:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c528729c8b344019216e12838463241\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/e9fc3f9ded8549f8a855ece6ee6367d1 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10abfbe573ca29ed1b12cba7fec8763960e8ad3b","revisions":{"10abfbe573ca29ed1b12cba7fec8763960e8ad3b":{"kind":"REWORK","_number":1,"created":"2023-02-09 11:36:40.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/44/873244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/44/873244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/873244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/873244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/44/873244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/44/873244/1"}}},"commit":{"parents":[{"commit":"e09f26a1dac9847140d6e7ea05d8ece145b73a84","subject":"Merge \"nodepool: infra-package-needs; cleanup tox installs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e09f26a1dac9847140d6e7ea05d8ece145b73a84"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-09 11:33:27.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-02-09 11:33:31.000000000","tz":330},"subject":"Cache Cirros 0.6.1 images","message":"Cache Cirros 0.6.1 images\n\n0.6.1 is the latest cirros release and with [1][2]\nis being used in neutron jobs.\n\nAdd these to nodepool images to avoid pulling it\nin jobs and hit external connectivity issues.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/869154\n[2] https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/869152\n\nChange-Id: Ic8b3e790fe332cf68bad7aaa3d5f85229600380b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10abfbe573ca29ed1b12cba7fec8763960e8ad3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10abfbe573ca29ed1b12cba7fec8763960e8ad3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
