)]}'
{"id":"openstack%2Fproject-config~873475","triplet_id":"openstack%2Fproject-config~master~I504682884a1439fac84d514880757c2cd041ada6","project":"openstack/project-config","branch":"master","topic":"farewell-limestone","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-02-13 21:24:41.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-13 22:05:29.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-02-13 22:05:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I504682884a1439fac84d514880757c2cd041ada6","subject":"Farewell limestone","status":"MERGED","created":"2023-02-13 00:07:20.000000000","updated":"2023-02-13 22:51:03.000000000","submitted":"2023-02-13 22:05:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"873475-farewell-limestone","meta_rev_id":"aa9947785a0c34efddb44ac682b140ec4377231e","_number":873475,"virtual_id_number":873475,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 22:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-13 21:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-13 21:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-02-13 21:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 00:14:46.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-02-13 00:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 21:13:56.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2023-02-13 21:26:31.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"4c58cef4c2c6448be8138e867193f6d163623f24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-13 00:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1ae89aad2c558496c4e60aeb3de5847fb868846","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-13 00:14:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"282f23f387bf53c0161625afe44fa165a3a9f359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-13 00:30:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"244c093e906768f08d90af482d7dff4601b8a05c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 00:55:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ce53f09fe94dbd9e14f6fdcfe1131d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4ba2b699244a4ed0a68648b213f85939 : SUCCESS in 23m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3196c2004d6f4befa181767920c5b3db : SUCCESS in 7m 02s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/1693bd082bb4432494095aa9d6d8fd30 : SUCCESS in 17m 36s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/10418066d88b4f068f16ced91760af31 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":2},{"id":"aeb3bf381a927c258a5a2d6c575f25bdcbf37c1a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-13 17:15:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08ac2a2de20822e72839557f50c14109a22abeef","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:13:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f01437faf85458728c987f1efd9f7932fca9b273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:22:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a10f9f8b379a088b1778222042abb65dc0e5f65","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:24:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f799e7ab5a0ca80c2cf45330c187b254b87881f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-13 21:24:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0fc0e079a5204ad795577f0b1babd79bdfbcf2d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:25:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a56c09bbb8839715ae4cdc1ae4103bba37ad8671","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:26:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3a74600a19da37eb038032512936283da2a20d70","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-13 21:26:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"165726a0e98c1598a1c2595c82eefbd3b2dba5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:48:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b906758b4b4752aff9570317b33c78\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/36a565316774443287fc5d394f496b42 : SUCCESS in 15m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa7aa3be69494066816ccced9e799037 : SUCCESS in 5m 17s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/c1d552587716470aa83f49dfe6986e8f : SUCCESS in 11m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/34dd416f596d4abfb60a7a0cd4ab84cf : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e2b7f4d754fc24a0837b13f54f442951d889626a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:48:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a6f22d050e37e1ad0fa70750cb7554a954c391c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 22:05:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3c6b255a4b1414aa515a9f16cf0fbe0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/05ce1a7f75c745358608b62833e439c9 : SUCCESS in 15m 17s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/804d0afd32bf43ae8ef4491841b26811 : SUCCESS in 6m 38s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/da4af3f8fa844429a9cc14710b7baa7a : SUCCESS in 11m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1f315ae3400d4a9f92cad172c0a91b4e : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2ac62f8f9c2e608c617ac055d576d701dc8ba535","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 22:05:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"aa9947785a0c34efddb44ac682b140ec4377231e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 22:51:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41dd11a187e4ed3a3201b785cc584fb\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/4228c03c9b614cf891a601e25dcd958e : SUCCESS in 11m 06s\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/248d264588c14e85a0e09cd386d672d5 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7c81cf6eda90d722a26ecc0197a749f5d1f31b60","revisions":{"c442b7980ab2820dafe105a807ae20592e033185":{"kind":"REWORK","_number":1,"created":"2023-02-13 00:07:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/75/873475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/873475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/873475/1"}}},"commit":{"parents":[{"commit":"d4b030bf50c2a919cbaa242b22d3b65fedbd37aa","subject":"Merge \"Add base openstack FIPS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4b030bf50c2a919cbaa242b22d3b65fedbd37aa"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:02:04.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:05:27.000000000","tz":0},"subject":"Farewell limestone","message":"Farewell limestone\n\nThe mirror in our Limestone Networks donor environment is now\nunreachable, but we ceased using this region years ago due to\npersistent networking trouble and the admin hasn\u0027t been around for\nroughly as long, so it\u0027s probably time to go ahead and say goodbye\nto it.\n\nIn preparation for cleanup of credentials in system-config, first\nremove configuration here except leave the nodepool provider with an\nempty diskimages list so that it will have a chance to pick up after\nitself.\n\nChange-Id: I504682884a1439fac84d514880757c2cd041ada6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c442b7980ab2820dafe105a807ae20592e033185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c442b7980ab2820dafe105a807ae20592e033185"}]},"branch":"refs/heads/master"},"c41a4a5ff747f1f91ffc9aa39b6be3b9c080514d":{"kind":"REWORK","_number":2,"created":"2023-02-13 00:30:47.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/75/873475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/873475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/873475/2"}}},"commit":{"parents":[{"commit":"d4b030bf50c2a919cbaa242b22d3b65fedbd37aa","subject":"Merge \"Add base openstack FIPS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4b030bf50c2a919cbaa242b22d3b65fedbd37aa"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:02:04.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:30:16.000000000","tz":0},"subject":"Farewell limestone","message":"Farewell limestone\n\nThe mirror in our Limestone Networks donor environment is now\nunreachable, but we ceased using this region years ago due to\npersistent networking trouble and the admin hasn\u0027t been around for\nroughly as long, so it\u0027s probably time to go ahead and say goodbye\nto it.\n\nIn preparation for cleanup of credentials in system-config, first\nremove configuration here except leave the nodepool provider with an\nempty diskimages list so that it will have a chance to pick up after\nitself.\n\nChange-Id: I504682884a1439fac84d514880757c2cd041ada6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c41a4a5ff747f1f91ffc9aa39b6be3b9c080514d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c41a4a5ff747f1f91ffc9aa39b6be3b9c080514d"}]},"branch":"refs/heads/master"},"2d067c76965a0f2f06a1ab70551ea20f0ea9852f":{"kind":"REWORK","_number":3,"created":"2023-02-13 21:22:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/75/873475/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/873475/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/873475/3"}}},"commit":{"parents":[{"commit":"d4b030bf50c2a919cbaa242b22d3b65fedbd37aa","subject":"Merge \"Add base openstack FIPS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4b030bf50c2a919cbaa242b22d3b65fedbd37aa"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:02:04.000000000","tz":0},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-13 21:22:14.000000000","tz":660},"subject":"Farewell limestone","message":"Farewell limestone\n\nThe mirror in our Limestone Networks donor environment is now\nunreachable, but we ceased using this region years ago due to\npersistent networking trouble and the admin hasn\u0027t been around for\nroughly as long, so it\u0027s probably time to go ahead and say goodbye\nto it.\n\nIn preparation for cleanup of credentials in system-config, first\nremove configuration here except leave the nodepool provider with an\nempty diskimages list so that it will have a chance to pick up after\nitself.\n\nChange-Id: I504682884a1439fac84d514880757c2cd041ada6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d067c76965a0f2f06a1ab70551ea20f0ea9852f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d067c76965a0f2f06a1ab70551ea20f0ea9852f"}]},"branch":"refs/heads/master"},"7c81cf6eda90d722a26ecc0197a749f5d1f31b60":{"kind":"REWORK","_number":4,"created":"2023-02-13 21:25:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/75/873475/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/873475/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/873475/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/873475/4"}}},"commit":{"parents":[{"commit":"d4b030bf50c2a919cbaa242b22d3b65fedbd37aa","subject":"Merge \"Add base openstack FIPS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d4b030bf50c2a919cbaa242b22d3b65fedbd37aa"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:02:04.000000000","tz":0},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-13 21:25:10.000000000","tz":660},"subject":"Farewell limestone","message":"Farewell limestone\n\nThe mirror in our Limestone Networks donor environment is now\nunreachable, but we ceased using this region years ago due to\npersistent networking trouble and the admin hasn\u0027t been around for\nroughly as long, so it\u0027s probably time to go ahead and say goodbye\nto it.\n\nIn preparation for cleanup of credentials in system-config, first\nremove configuration here except leave the nodepool provider with an\nempty diskimages list so that it will have a chance to pick up after\nitself.\n\nChange-Id: I504682884a1439fac84d514880757c2cd041ada6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c81cf6eda90d722a26ecc0197a749f5d1f31b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c81cf6eda90d722a26ecc0197a749f5d1f31b60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
