)]}'
{"id":"openstack%2Fproject-config~873476","triplet_id":"openstack%2Fproject-config~master~I69b96b690918a9145d2e7ccbc79968c5341480bb","project":"openstack/project-config","branch":"master","topic":"farewell-limestone","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-02-13 23:15:33.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-02-13 23:25:18.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-02-13 23:25:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69b96b690918a9145d2e7ccbc79968c5341480bb","subject":"Remove empty limestone nodepool providers","status":"MERGED","created":"2023-02-13 00:07:20.000000000","updated":"2023-02-13 23:59:22.000000000","submitted":"2023-02-13 23:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873476-farewell-limestone","meta_rev_id":"63df16b5dfb472d844d03a03eec817bbd9855403","_number":873476,"virtual_id_number":873476,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 23:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-13 23:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-13 21:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-02-13 23:15:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 00:15:00.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-02-13 00:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 21:26:45.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"97f9d275b7efe212e5feda68e921de57833a65c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-13 00:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40150c44099123e71fdcbfb2b964c0db5eb919bc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-13 00:08:08.000000000","message":"Set Work In Progress\n\nHold until the parent change has deployed and no images remain in Glance.","accounts_in_message":[],"_revision_number":1},{"id":"fd6170b4aafc931b2a760d89b0d2f12747d1f1c0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-13 00:15:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"062f061438330a89a220fc04cb123dac31480ca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 00:23:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e5b0e395444c7db83b62be415e5257\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d4759e2236db4ed3bf492a1221c3c62c : SUCCESS in 2m 37s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/94d78ab3ceab4b35989f3da59f5e11fd : SUCCESS in 5m 05s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8da62b32dc08401983c5b2ea9664e1c6 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"feb8b2f893fdb47d14b27a0a7dcddd7b959d2dbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-13 00:30:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"7e4f9ebc74b7d2e97d2464579d9a293bace673b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 00:39:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2971c14a589248f0b4c235e843cf1e75\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fb6b99f91d644a4b98d617bcf95a486d : SUCCESS in 3m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ef81cc266f9a4ec09fa24303414384ee : SUCCESS in 7m 38s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ccd3ff10ca6648f1a9ade97d920ac7e2 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2a4e71db09b99e7e276cff46779d76a856343da4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:22:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"914f613f74ba8a3727081bc50bc0f487d39a7928","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:25:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a00190c462f1df2820d124f6da001f034146eb65","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-02-13 21:26:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e9e56cd79fcd978e09dff26e6dde233a7d09e9d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:38:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02e0ff34fab44a228e68a7047e662eda\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bc922f8a9a4f44808b86837537099f9b : SUCCESS in 3m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/25c66df6981e4b3aad3050f185e2815e : SUCCESS in 6m 06s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4ff71e0d1fe84faf90c63daf38360e7e : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":4},{"id":"1256a746f625c8963a5ac86aa574fc484f188ce3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-13 23:08:01.000000000","message":"Set Ready For Review\n\nThe parent has deployed now.","accounts_in_message":[],"_revision_number":4},{"id":"42b7f9e21a05924eeb5325f150aadec30b4e87b8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-13 23:15:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0d3ba51e77ea8f2ccbc0814e830d650ba4a3928b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:15:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4b6cce569ad7a96ef6977f2e789811a183e2e979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:25:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843bd9e540684cd98375c9490c9a423d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e23f4aad8f1941b2a7538be8af38948e : SUCCESS in 6m 23s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/45d9766b5536429e8ca8e6fcaf95383f : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"80af15d5eb9e2a3b16f289f74476174c78b702d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:25:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"63df16b5dfb472d844d03a03eec817bbd9855403","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:59:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b752029fac84742a5ff5087acf72649\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/b8fc4ed576da4f589cef8b51f13f90f1 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"92814f9b71789e05e698f561431204705c49d41a","revisions":{"2dfad4aa212b7c5285003f0a15cccbcb60b18b57":{"kind":"REWORK","_number":1,"created":"2023-02-13 00:07:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/76/873476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/873476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/873476/1"}}},"commit":{"parents":[{"commit":"c442b7980ab2820dafe105a807ae20592e033185","subject":"Farewell limestone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c442b7980ab2820dafe105a807ae20592e033185"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:06:05.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:06:05.000000000","tz":0},"subject":"Remove empty limestone nodepool providers","message":"Remove empty limestone nodepool providers\n\nOnce the builders have a chance to clear out all uploaded images,\nthis will remove the remaining references in Nodepool. Then\nsystem-config cleanup can proceed.\n\nChange-Id: I69b96b690918a9145d2e7ccbc79968c5341480bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2dfad4aa212b7c5285003f0a15cccbcb60b18b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2dfad4aa212b7c5285003f0a15cccbcb60b18b57"}]},"branch":"refs/heads/master"},"73ccefc07de87c3f5bea3457ade408dcce3cf7f9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-13 00:30:47.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/76/873476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/873476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/873476/2"}}},"commit":{"parents":[{"commit":"c41a4a5ff747f1f91ffc9aa39b6be3b9c080514d","subject":"Farewell limestone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c41a4a5ff747f1f91ffc9aa39b6be3b9c080514d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:06:05.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:30:17.000000000","tz":0},"subject":"Remove empty limestone nodepool providers","message":"Remove empty limestone nodepool providers\n\nOnce the builders have a chance to clear out all uploaded images,\nthis will remove the remaining references in Nodepool. Then\nsystem-config cleanup can proceed.\n\nChange-Id: I69b96b690918a9145d2e7ccbc79968c5341480bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/73ccefc07de87c3f5bea3457ade408dcce3cf7f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/73ccefc07de87c3f5bea3457ade408dcce3cf7f9"}]},"branch":"refs/heads/master"},"95be92e1f2d0b9f747196c5d6d6f0dc6f2abaea8":{"kind":"REWORK","_number":3,"created":"2023-02-13 21:22:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/76/873476/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/873476/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/873476/3"}}},"commit":{"parents":[{"commit":"2d067c76965a0f2f06a1ab70551ea20f0ea9852f","subject":"Farewell limestone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d067c76965a0f2f06a1ab70551ea20f0ea9852f"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:06:05.000000000","tz":0},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-13 21:22:41.000000000","tz":660},"subject":"Remove empty limestone nodepool providers","message":"Remove empty limestone nodepool providers\n\nOnce the builders have a chance to clear out all uploaded images,\nthis will remove the remaining references in Nodepool. Then\nsystem-config cleanup can proceed.\n\nChange-Id: I69b96b690918a9145d2e7ccbc79968c5341480bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95be92e1f2d0b9f747196c5d6d6f0dc6f2abaea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/95be92e1f2d0b9f747196c5d6d6f0dc6f2abaea8"}]},"branch":"refs/heads/master"},"92814f9b71789e05e698f561431204705c49d41a":{"kind":"REWORK","_number":4,"created":"2023-02-13 21:25:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/76/873476/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/873476/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/873476/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/873476/4"}}},"commit":{"parents":[{"commit":"7c81cf6eda90d722a26ecc0197a749f5d1f31b60","subject":"Farewell limestone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7c81cf6eda90d722a26ecc0197a749f5d1f31b60"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-13 00:06:05.000000000","tz":0},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-02-13 21:25:25.000000000","tz":660},"subject":"Remove empty limestone nodepool providers","message":"Remove empty limestone nodepool providers\n\nOnce the builders have a chance to clear out all uploaded images,\nthis will remove the remaining references in Nodepool. Then\nsystem-config cleanup can proceed.\n\nChange-Id: I69b96b690918a9145d2e7ccbc79968c5341480bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/92814f9b71789e05e698f561431204705c49d41a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/92814f9b71789e05e698f561431204705c49d41a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
