)]}'
{"id":"openstack%2Fproject-config~873843","triplet_id":"openstack%2Fproject-config~master~Ic07f6e00abfa4dcc546dd663d4458debdcaf707f","project":"openstack/project-config","branch":"master","topic":"gerrit-3.7","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-04-06 23:47:24.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-04-06 23:47:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic07f6e00abfa4dcc546dd663d4458debdcaf707f","subject":"Rename x/xstatic-angular-fileupload-\u003eopenstack/xstatic-angular-fileupload","status":"MERGED","created":"2023-02-15 06:43:38.000000000","updated":"2023-04-06 23:49:04.000000000","submitted":"2023-04-06 23:47:24.000000000","submitter":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873843-gerrit-3.7","meta_rev_id":"7ce0ad317aea4e2a1bd905f3944e3d1f6e767179","_number":873843,"virtual_id_number":873843,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-04-06 23:47:24.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-04 22:22:41.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-04-05 20:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-04-06 23:47:24.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-04 22:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-04-06 17:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2023-04-06 23:47:24.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}],"CC":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 06:43:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-15 06:54:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 23:03:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-02-16 02:01:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-04-04 22:23:27.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2023-04-05 20:43:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-04-06 17:48:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-04-06 23:47:24.000000000","updated_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"reviewer":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"state":"REVIEWER"}],"messages":[{"id":"32f161aee22ee6329a339dd3cb7067c77e628564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 06:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"672bb6ece22cc2199eef1e673faf23b577640a58","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-15 06:43:57.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9d406c8953d21d8e968ca4886933275b64604b2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 06:54:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fadcdb73ce325a51efcc27fd3b90d05f738b11c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-15 06:54:43.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"7193f839464075c288ad8fa87b95a2bc722baaa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 06:54:58.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/xstatic-angular-fileupload\" was not found.  All\n  projects referenced within a Zuul configuration must first be added to\n  the main configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/xstatic-angular-fileupload\n      templates:\n        - noop-jobs\n        - publish-xstatic-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5253, column 3","accounts_in_message":[],"_revision_number":2},{"id":"2d7d6f9a9aa6dc1cf283c419242982a02c5f7e1a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-15 07:01:20.000000000","message":"Topic set to project-rename","accounts_in_message":[],"_revision_number":2},{"id":"cabcd1d91ec252c16f44ea919023411f35016abd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-15 23:03:25.000000000","message":"Patch Set 2:\n\nPlease add this to https://wiki.openstack.org/wiki/Meetings/InfraTeamMeeting#Upcoming_Project_Renames and we\u0027ll discuss scheduling a maintenance window for it in the next weekly meeting (2023-02-28 since we\u0027re skipping the next one due to absences).","accounts_in_message":[],"_revision_number":2},{"id":"f06c97e2f80357bbff64e82d56fee569408e16d4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-16 02:01:26.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32ee65c75bd9623ada603f0626c3900a8772a266","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-16 05:19:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88e0b7a2c3e4d5a73edb93059e4e034a68dffe67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-04 22:22:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"86e6fa5f0458c48d63239fc9df7c3ccb3fd0003d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-04 22:22:32.000000000","message":"Patch Set 3:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"5101744ede997256691e859b64e58a84d9b3fc01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 22:22:41.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"25eabe76815e9cb59d66146d953c1a55b5057d4b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-04 22:23:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10952cc9b209e7a6491fa111a70c85a8a617a96c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-05 20:43:50.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f9f410c8031293fdebcd8190d32d5074f3466986","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-06 17:48:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a86c9a9fbbbc750150f6066ec7801e372f6c4e4e","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2023-04-06 23:47:24.000000000","message":"Patch Set 3: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul for rename","accounts_in_message":[],"_revision_number":3},{"id":"ebd560117a33d8cf15b7acb9d66cd2359c217de1","tag":"autogenerated:gerrit:merged","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2023-04-06 23:47:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7ce0ad317aea4e2a1bd905f3944e3d1f6e767179","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 23:49:04.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/virtualpdu\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/virtualpdu\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5180, column 3","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62f9137743c1856c6484eee3a718637c88090f0c","revisions":{"52d9d4c3072c0e51727f078d2529728eed1fcf15":{"kind":"REWORK","_number":1,"created":"2023-02-15 06:43:38.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/43/873843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/873843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/873843/1"}}},"commit":{"parents":[{"commit":"3e52d32877bcc0ea6276e9ca98af1bf342002b68","subject":"Merge \"Cache Cirros 0.6.1 images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e52d32877bcc0ea6276e9ca98af1bf342002b68"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 06:35:24.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 06:43:34.000000000","tz":330},"subject":"Rename x/xstatic-angular-fileupload to openstack/xstatic-angular-fileupload","message":"Rename x/xstatic-angular-fileupload to openstack/xstatic-angular-fileupload\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n\nChange-Id: Ic07f6e00abfa4dcc546dd663d4458debdcaf707f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52d9d4c3072c0e51727f078d2529728eed1fcf15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/52d9d4c3072c0e51727f078d2529728eed1fcf15"}]},"branch":"refs/heads/master"},"2ae82405be6eed8d22a099a3f99207d23e804923":{"kind":"REWORK","_number":2,"created":"2023-02-15 06:54:25.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/43/873843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/873843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/873843/2"}}},"commit":{"parents":[{"commit":"3e52d32877bcc0ea6276e9ca98af1bf342002b68","subject":"Merge \"Cache Cirros 0.6.1 images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e52d32877bcc0ea6276e9ca98af1bf342002b68"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 06:35:24.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 06:53:08.000000000","tz":330},"subject":"Rename x/xstatic-angular-fileupload-\u003eopenstack/xstatic-angular-fileupload","message":"Rename x/xstatic-angular-fileupload-\u003eopenstack/xstatic-angular-fileupload\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n\nChange-Id: Ic07f6e00abfa4dcc546dd663d4458debdcaf707f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2ae82405be6eed8d22a099a3f99207d23e804923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2ae82405be6eed8d22a099a3f99207d23e804923"}]},"branch":"refs/heads/master"},"62f9137743c1856c6484eee3a718637c88090f0c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-04 22:22:05.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/873843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/873843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/873843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/873843/3"}}},"commit":{"parents":[{"commit":"f2d80da134b0908170cde961fe49f444485336a0","subject":"Ironic program adopting virtualpdu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2d80da134b0908170cde961fe49f444485336a0"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-15 06:35:24.000000000","tz":330},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-04 22:21:09.000000000","tz":600},"subject":"Rename x/xstatic-angular-fileupload-\u003eopenstack/xstatic-angular-fileupload","message":"Rename x/xstatic-angular-fileupload-\u003eopenstack/xstatic-angular-fileupload\n\nxstatic-angular-fileupload pypi is used by Horizon.\nHorizon team discussed this topic in the horizon weekly\nmeeting and decided to move it under the horizon deliverables\nand keep them under the OpenStack namespace [1].\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-horizon/%23openstack-horizon.2023-02-01.log.html#t2023-02-01T16:09:57\n\nChange-Id: Ic07f6e00abfa4dcc546dd663d4458debdcaf707f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62f9137743c1856c6484eee3a718637c88090f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62f9137743c1856c6484eee3a718637c88090f0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
