)]}'
{"id":"openstack%2Fproject-config~875291","triplet_id":"openstack%2Fproject-config~master~I8754673dc386f5f2f649235b26c68b34ec8ccf8d","project":"openstack/project-config","branch":"master","topic":"retire-puppet-tacker","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2023-02-27 07:50:08.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-05-02 13:57:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8754673dc386f5f2f649235b26c68b34ec8ccf8d","subject":"Retire puppet-tacker - Step 5: Remove Project","status":"MERGED","created":"2023-02-25 05:53:18.000000000","updated":"2023-05-02 14:15:33.000000000","submitted":"2023-05-02 13:57:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875291-retire-puppet-tacker","meta_rev_id":"ab61e27794111773cd5387dd61aef256ed861055","_number":875291,"virtual_id_number":875291,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-02 13:57:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2023-02-27 07:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-27 20:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-05-02 11:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2023-05-02 11:59:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-25 05:53:35.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-25 05:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 07:46:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2023-02-27 07:46:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2023-02-27 20:33:57.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-05-02 11:59:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e368e17067a901ead12e95d29487fba9a8a3abdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-25 05:53:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e26a3fde768997ae5c0368be2f809e8dc87ebb0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-25 05:53:35.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"07eb8ae5e9b1014dcb3202727beb805dae866852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 05:53:41.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"912f8ca26c8dc837fd6082e587ffd719f135def0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-25 06:02:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f89aa1bfb74b5f1b57b12a016702a5564cc92753","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-25 06:02:58.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"7d130ab6c4e2b1d8474851585ff6e023478eb905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 06:03:07.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"68ce39ca077ac477bdfa710889c7c6a5c357848f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-25 06:04:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5e19754a217d624d03ec307fb621ce1766f24461","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-25 06:04:19.000000000","message":"Patch Set 3:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"cca9feeb6288931785356ab59ea5a3116ee8e60a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 06:04:22.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"e07efac6bc95898642038bfd803d80b1bd0de509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-27 02:24:54.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5637ae293b641086dd118c23879b42e1851a0fd5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-27 02:25:20.000000000","message":"Patch Set 4:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"9e7eb4da9dba7e99dca828806f89e3ad3ab90f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 02:56:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d724fe00557496c9ae65eb4015b7f48\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6e7ed178eaf24ed2b4a2d8761e5ccd22 : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/326736b32b964358898d9f37d32e8ab0 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f2d1ea857446476e89cffc2637633cf1 : SUCCESS in 3m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/35802be54c354abdbb9baf8e25a4fbc0 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0cd45764b884c4be0312069f712b8829fe19a137","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-02-27 07:46:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b342bdf67c9f7a694915bdadb9feed0ab5af20d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-02-27 07:46:23.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"6cc969da58e301930fc3f29594b62e7cf70ed41a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-27 07:47:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d9d32846eee9e7ed4547007717816b0772a0853","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-27 07:48:08.000000000","message":"Patch Set 5:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":5},{"id":"9b921632149425a12b168bddafff0c1fc3e24f5d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-02-27 07:48:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b9f416f494e21e36975c364723f264239b29976f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-27 07:49:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f70961cdf439c34eb07dc4f2cca5f501ef854302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 08:01:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37585875e9ad409ba5be7eb7e2d66b23\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2fd7803a39d9405092e614f0635cfc05 : SUCCESS in 2m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cb0b4f1afd3a4e05be8c8c9ba3f88840 : SUCCESS in 6m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f87b500703614bc9836386647d327611 : SUCCESS in 3m 21s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8d4c5a446be643108b2ab8f0b4b0fc01 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":5},{"id":"7a355f586a550e20e314db023c5097836865e089","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-27 20:33:57.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3bfc7cbd89e8d53584f122ef023d68c29de3fac4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 11:59:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e667da4e2df628336fb525820afb4c7cb5b671bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 12:00:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e8fadbcceb1980e0aa90f0f67ad03fe2715b12a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 12:26:24.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8282536fc1b4810912580d969d8cd63\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6b61b234818f45fa874cc325e4122e23 : FAILURE in 5m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/608bbb5bae034e1b9cfde982f1d433d0 : SUCCESS in 3m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/6bb7ff1ba0264c64bd329ae6d411e41f : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":5},{"id":"03b8d82c2d58a79010c4a53fe92844ffa5950e12","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 12:34:45.000000000","message":"Patch Set 5:\n\nThe linters error seems to be the same one we pinned ansible-compat for in https://review.opendev.org/881938 a few hours ago.","accounts_in_message":[],"_revision_number":5},{"id":"6c85a05af0833ccbb673b71d1c3434971d9816eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-02 12:35:06.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":5},{"id":"a1cbfbb1763fcee24a7936200ea00d8d2ed0a4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 12:43:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4c5f666293444858c45ef2352aab4da\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/90469754084e41549846b6cbfe1ba05e : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d974d03a635b4c2a9b592adc63c62e92 : FAILURE in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b831e271db954bc49056f4eac2193948 : SUCCESS in 3m 39s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/a8ba21d0ff9a49c0b4cd654384eb4955 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":5},{"id":"59973565aecb81bfcfabe826f6e6c6714df8a060","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 13:32:39.000000000","message":"Patch Set 5:\n\nrecheck now that https://review.opendev.org/881963 has merged to address the linters failure","accounts_in_message":[],"_revision_number":5},{"id":"c9ace58100d8a6bf95397eba7a5a3985ce2b4cd7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-02 13:32:58.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":5},{"id":"6f8f22e8476861dc218e039090d344b028793da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:44:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/459e014d2c634bc5ad6378583b68a653\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/81f833840cbc4b79bf5644ca5f5789fd : SUCCESS in 3m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0047ea3dda7b410f8148e3274259857d : SUCCESS in 6m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a48497cba393492c8424ddc8caefb31a : SUCCESS in 3m 09s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c3a56625a93a4fbb928b8fa3c83ada9e : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":5},{"id":"90709c3fdb69b48025b6b752e925cbadc928fa31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:45:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"325b88fe9ae8da4a2a20e2600310676d16ba1e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:57:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20814f53291b4a94bd7777a551b49d69\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3856ee8b2ea246b9ae5d59534ca31aaa : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cfbb200faab44921b8a7f47a62610fa6 : SUCCESS in 3m 40s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/092d7fcee30c42b482a579ad44e6e6c7 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":5},{"id":"f98d8ebd19483c1312953a0e77036bf8fb559b1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:57:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ab61e27794111773cd5387dd61aef256ed861055","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:15:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fba440c7a0274df19cf602645be403b8\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1b2e6a9c0430463f9d38775574095409 : SUCCESS in 7m 21s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6a089b4e5ef845c5a62e6e69a3d5ed7c : SUCCESS in 5m 50s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/225ba3c1e3d246998a7fd6f9411525f7 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"335fc153ddeec2ad167cd8a6660fba8d44d269c9","revisions":{"9b0f2231ff663dfa5b3eaafe0968c8e4da091e5a":{"kind":"REWORK","_number":1,"created":"2023-02-25 05:53:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/875291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/875291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/875291/1"}}},"commit":{"parents":[{"commit":"e3131db576ff1439dcc1dd4a83b90bcfe8f50962","subject":"Retire puppet-tacker - Step 1: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3131db576ff1439dcc1dd4a83b90bcfe8f50962"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 05:52:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 05:52:42.000000000","tz":540},"subject":"Retire puppet-tacker - Step 3: Retire Repository","message":"Retire puppet-tacker - Step 3: Retire Repository\n\nDepends-on: https://review.opendev.org/875290\nChange-Id: I8754673dc386f5f2f649235b26c68b34ec8ccf8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b0f2231ff663dfa5b3eaafe0968c8e4da091e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b0f2231ff663dfa5b3eaafe0968c8e4da091e5a"}]},"branch":"refs/heads/master"},"0ce75b66c05362a61eb8cd7ee0c709d742170194":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-25 06:02:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/875291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/875291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/875291/2"}}},"commit":{"parents":[{"commit":"e3131db576ff1439dcc1dd4a83b90bcfe8f50962","subject":"Retire puppet-tacker - Step 1: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3131db576ff1439dcc1dd4a83b90bcfe8f50962"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 05:52:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 06:02:40.000000000","tz":0},"subject":"Retire puppet-tacker - Steps 5: Remove Project","message":"Retire puppet-tacker - Steps 5: Remove Project\n\nDepends-on: https://review.opendev.org/875295\nChange-Id: I8754673dc386f5f2f649235b26c68b34ec8ccf8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ce75b66c05362a61eb8cd7ee0c709d742170194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ce75b66c05362a61eb8cd7ee0c709d742170194"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0aa04d4536199c033169966a7517638db23f5029":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-25 06:04:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/875291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/875291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/875291/3"}}},"commit":{"parents":[{"commit":"e3131db576ff1439dcc1dd4a83b90bcfe8f50962","subject":"Retire puppet-tacker - Step 1: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3131db576ff1439dcc1dd4a83b90bcfe8f50962"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 05:52:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 06:04:01.000000000","tz":0},"subject":"Retire puppet-tacker - Step 5: Remove Project","message":"Retire puppet-tacker - Step 5: Remove Project\n\nDepends-on: https://review.opendev.org/875295\nChange-Id: I8754673dc386f5f2f649235b26c68b34ec8ccf8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0aa04d4536199c033169966a7517638db23f5029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0aa04d4536199c033169966a7517638db23f5029"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a16f977f7baca236838541a2dbd56b5221b11905":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-27 02:24:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/875291/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/875291/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/875291/4"}}},"commit":{"parents":[{"commit":"875dbd792462a80f7b861a8c2a9bbf8ff194d2cf","subject":"Retire puppet-tacker - Step 1: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/875dbd792462a80f7b861a8c2a9bbf8ff194d2cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 05:52:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-27 02:24:54.000000000","tz":0},"subject":"Retire puppet-tacker - Step 5: Remove Project","message":"Retire puppet-tacker - Step 5: Remove Project\n\nDepends-on: https://review.opendev.org/875295\nChange-Id: I8754673dc386f5f2f649235b26c68b34ec8ccf8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a16f977f7baca236838541a2dbd56b5221b11905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a16f977f7baca236838541a2dbd56b5221b11905"}]},"branch":"refs/heads/master","description":"Rebase"},"335fc153ddeec2ad167cd8a6660fba8d44d269c9":{"kind":"REWORK","_number":5,"created":"2023-02-27 07:47:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/875291/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/875291/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/875291/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/875291/5"}}},"commit":{"parents":[{"commit":"875dbd792462a80f7b861a8c2a9bbf8ff194d2cf","subject":"Retire puppet-tacker - Step 1: End project Gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/875dbd792462a80f7b861a8c2a9bbf8ff194d2cf"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-25 05:52:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-27 07:47:36.000000000","tz":540},"subject":"Retire puppet-tacker - Step 5: Remove Project","message":"Retire puppet-tacker - Step 5: Remove Project\n\nDepends-on: https://review.opendev.org/875295\nChange-Id: I8754673dc386f5f2f649235b26c68b34ec8ccf8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/335fc153ddeec2ad167cd8a6660fba8d44d269c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/335fc153ddeec2ad167cd8a6660fba8d44d269c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
