)]}'
{"id":"openstack%2Fproject-config~875495","triplet_id":"openstack%2Fproject-config~master~I480d695a63f0a695631c97294740c8443dd6981c","project":"openstack/project-config","branch":"master","topic":"disable-inmotion","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-02-27 16:42:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I480d695a63f0a695631c97294740c8443dd6981c","subject":"Revert \"Temporarily stop booting nodes in inmotion iad3\"","status":"MERGED","created":"2023-02-27 13:43:34.000000000","updated":"2023-02-27 16:52:44.000000000","submitted":"2023-02-27 16:42:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875495-disable-inmotion","meta_rev_id":"e7a5f8460743dda79582a47f883d7c922ad2a217","_number":875495,"virtual_id_number":875495,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-27 16:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-02-27 16:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-02-27 16:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 13:55:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 16:31:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"a978cc32b82c5ba3b4e68681e847c4cfb89d7e1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-27 13:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9c00edfc67e624ccf0b32fa769a0c0657fc2db7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:55:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e65acbb231451f8c65323d3704cb6e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c36ee4e71ebc434c8a6abca07d3f8d5e : SUCCESS in 2m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3e193215416c4cb88beef26184957026 : SUCCESS in 5m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/af9ed0b7b8f54502a650fe2553aaa965 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8e961f698e85909bfdc2db7fb6e1953c0068ff2b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-02-27 13:55:37.000000000","message":"Set Ready For Review\n\nThe server has been booted and both AFS and Apache mod_proxy caches have been tested successfully with a web browser.","accounts_in_message":[],"_revision_number":1},{"id":"b3ea729cde7a63d63857deed66aa89c2e13fb63b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-02-27 16:31:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"029b2d6b05bddefd449e5a5959ffd6dd7dbbe15e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:31:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4656c44a92e1e5346aafe0da92096e8f714a0dc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:42:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef40578d36b491c9d151e5397a35c52\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/efc6e75619a84846a1f11b4c1b86779c : SUCCESS in 7m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/19a7178dc13d4f1ea1287d09ebb89540 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6fd4360164b8455925ad80163c884dc6cdf357a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:42:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e7a5f8460743dda79582a47f883d7c922ad2a217","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/335a23bee8cb4281ace4faad41d78e31\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/1e9f4d53701340828409aedec535608f : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f45f51fdd726462d16441668ec5ac1a2e6d6c0be","revisions":{"f45f51fdd726462d16441668ec5ac1a2e6d6c0be":{"kind":"REWORK","_number":1,"created":"2023-02-27 13:43:34.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/95/875495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875495/1"}}},"commit":{"parents":[{"commit":"17888a4a0352e16468384f3198ba2ba1fd0483b3","subject":"Temporarily stop booting nodes in inmotion iad3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17888a4a0352e16468384f3198ba2ba1fd0483b3"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-27 13:42:50.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-02-27 13:43:27.000000000","tz":0},"subject":"Revert \"Temporarily stop booting nodes in inmotion iad3\"","message":"Revert \"Temporarily stop booting nodes in inmotion iad3\"\n\nMerge once the mirror for this provider has returned to service.\n\nThis reverts commit 17888a4a0352e16468384f3198ba2ba1fd0483b3.\n\nChange-Id: I480d695a63f0a695631c97294740c8443dd6981c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f45f51fdd726462d16441668ec5ac1a2e6d6c0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f45f51fdd726462d16441668ec5ac1a2e6d6c0be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
