)]}'
{"id":"openstack%2Fproject-config~875993","triplet_id":"openstack%2Fproject-config~master~I83160aeec0a450f8678ecb583fb7570ac0e71f4a","project":"openstack/project-config","branch":"master","topic":"gerrit-s-r-3.7","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-03-13 21:26:01.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I83160aeec0a450f8678ecb583fb7570ac0e71f4a","subject":"gerrit/acl : Convert Backport-Candidate to submit-requirements","status":"MERGED","created":"2023-03-02 00:18:18.000000000","updated":"2023-03-13 21:44:37.000000000","submitted":"2023-03-13 21:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875993-gerrit-s-r-3.7","meta_rev_id":"1040544975cfd36599ec6be89da34757fd1a9f08","_number":875993,"virtual_id_number":875993,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 21:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-03-07 23:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-13 21:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 00:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 17:20:37.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2023-03-07 23:07:10.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"e76fb295ec2768fdfe9ed2a535eb40b7e1058d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 00:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c04344b3718c51694d0ca30ec86439e2a0ea065d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 00:30:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6a9a2f164c499b94bc004b46f8d66f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2eb6c765cc464726a90d7b5a53938c04 : SUCCESS in 2m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/917dff2199b149baa0e5aaf70e7ec8d8 : SUCCESS in 6m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/38387fc9451647f8b2cbde793c412205 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"58ae7f25d5f022a81eb4773f6d544d7672b7f200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 05:48:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b58478743f71ae9a864e72cfeac2d0ee012e1cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 05:56:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c30bb5a694e4f419a0759efac09feec\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7cc35d446f664f429066573304052283 : SUCCESS in 2m 40s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/23719ba1ae274448a24ad7a1344d711b : SUCCESS in 5m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/de5d03638849437bb249e24594a516fb : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"80f84bed2ade6e1b3c1a3af3d1bf584ff3ed8817","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 03:00:01.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":2},{"id":"350ae7be0f57ab756450d04ac423300c249ca7da","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 03:00:03.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":2},{"id":"da94b822c6fd63bd25f357bbbce371a761c53ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 03:42:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c7072eff3ba3f39c877b4ec7eab143758d102670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:14:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/445396649e424522b14b365e650352de\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7d6f82a938bb49858e3ecbdbedef12f4 : SUCCESS in 3m 11s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63686ac547274cfb9a2b0df70dde0f35 : SUCCESS in 6m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3dd7be7bce9646a5b52e6c536d3e763b : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"6e4d526919d26da05873c4865ffc6b5d97585b80","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-06 17:20:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce23b2eaf489e9746bab1636cea2f31be647c801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 04:03:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"695a53174514736a411724cf3b15cb9fc12d30ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 04:11:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca60b144dc4461b960022f91e65a1a0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/74f759a1b12f497da33aae4e8477fe36 : SUCCESS in 2m 46s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f94709bebab4b1eb29d62008efc6366 : SUCCESS in 5m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/40538cd4fd3f47e68ac917d8565bab74 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"1732339fd92e3aa79b5229f7610998c466a5918d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:17:52.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":4},{"id":"595a62e750d3f63a9c53de2406c3a2dbdfaed0c6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:17:58.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":4},{"id":"6184e7d5f4966b8feddc139bb2390a346a7e1704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 22:21:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9234538900cfc2a697cd29548f1212ab0c6ee62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:39:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef4f28d6f114e72bf3f4d5859bd66ee\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1a7081d507f140bf97a7ad75372fba1f : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/083735783e824ea18f51fe44a9086965 : SUCCESS in 5m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8457255edc564f8eb73937c9db401ebb : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":5},{"id":"85e1f4d19bb5801586b6d34a384625a5d81a0506","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-07 23:07:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c89e1710e0081647f557bf82091ed245dcfc7bba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 23:13:16.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"591a4bc7b1bc049e7f177b817c93d656872924ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 23:20:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72ff79ab4b1b45f8a6edd5426820dfc6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fe7e9aad9155472e855f9a5b5822a89c : SUCCESS in 3m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/30f7e3e870e64a8f8c735f61d1cf7506 : SUCCESS in 6m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/db52db6be4dc449998627816cef17f16 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":6},{"id":"1422fc652fc83eac4461a26e4187e39a04c2c3f0","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-13 21:26:01.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1506cf5d94ff799a0a8840ec50961b2bdec7209f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 21:26:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e618e9670264e37a1207874ae44be2cb5d9aefe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 21:32:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3352856a25484b8f9dc662269e30ccea\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fd9496f258b9488d91ca49e4e639f181 : SUCCESS in 4m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/20349da0aff24ae6b312e2bf2d0c5d2b : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":6},{"id":"ab92a34a81bf76c009f2e2d9aa69ccb39fcaeb71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 21:32:15.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"1040544975cfd36599ec6be89da34757fd1a9f08","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 21:44:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7142b467e57d4f26b68100e274609d65\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/53d1f0b6008e48d78cc618be92d84ff8 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"69c2976cab6cfe860a35e8a63e6cdf0177fe592c","revisions":{"2c4b0009742f81ca8024c5a4d948dfa18a2943f8":{"kind":"REWORK","_number":1,"created":"2023-03-02 00:18:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/875993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/875993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/875993/1"}}},"commit":{"parents":[{"commit":"fad60cc7ccd655b91326d4724ddebec65df7d3e3","subject":"gerrit/acl : remove deprecated NoOp function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fad60cc7ccd655b91326d4724ddebec65df7d3e3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"subject":"gerrit/acl : convert NoBlock to submit-requirements","message":"gerrit/acl : convert NoBlock to submit-requirements\n\nNoBlock is the same as NoOp (converted with prior change\nIc43f561174ebf30474b1b54be2bed02695cebedc) but it looks like we\u0027ve\ncopy-pasted this for the backport flag.\n\nThis updates all NoBlock ACLs to use a submit-requirement of \"is:true\"\nwhich should be the same no-op semantics.\n\nChange-Id: I83160aeec0a450f8678ecb583fb7570ac0e71f4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c4b0009742f81ca8024c5a4d948dfa18a2943f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c4b0009742f81ca8024c5a4d948dfa18a2943f8"}]},"branch":"refs/heads/master"},"e8b6fc0bb244d3f8c40e58374c2e599abd979be5":{"kind":"REWORK","_number":2,"created":"2023-03-02 05:48:20.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/875993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/875993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/875993/2"}}},"commit":{"parents":[{"commit":"111683726ed4661b268f7e27b4d32f849eda3569","subject":"gerrit/acl : submit-requirements for deprecated NoOp function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/111683726ed4661b268f7e27b4d32f849eda3569"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 05:16:08.000000000","tz":660},"subject":"gerrit/acl : add submit requirements to NoBlock labels","message":"gerrit/acl : add submit requirements to NoBlock labels\n\nNoBlock is the same as NoOp (submit requirements added with prior\nchange Ic43f561174ebf30474b1b54be2bed02695cebedc) but it looks like\nwe\u0027ve copy-pasted this for the backport flag.  For consistency, change\nthis to NoOp (a later change will lint this).\n\nAs the prior change, this updates all NoBlock ACLs to use a\nsubmit-requirement of \"is:true\" which should be the same no-op\nsemantics.\n\nChange-Id: I83160aeec0a450f8678ecb583fb7570ac0e71f4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8b6fc0bb244d3f8c40e58374c2e599abd979be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8b6fc0bb244d3f8c40e58374c2e599abd979be5"}]},"branch":"refs/heads/master"},"5efd530486af8659e81dd55a8980646ce6d9c9aa":{"kind":"REWORK","_number":3,"created":"2023-03-03 03:42:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/875993/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/875993/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/875993/3"}}},"commit":{"parents":[{"commit":"593fa4f543bb47c1c13f8bf575fe557973ef619b","subject":"gerrit/acl : submit-requirements for deprecated NoOp function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/593fa4f543bb47c1c13f8bf575fe557973ef619b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-03 03:39:22.000000000","tz":660},"subject":"gerrit/acl : add submit requirements to NoBlock labels","message":"gerrit/acl : add submit requirements to NoBlock labels\n\nNoBlock is the same as NoOp (submit requirements added with prior\nchange Ic43f561174ebf30474b1b54be2bed02695cebedc) but it looks like\nwe\u0027ve copy-pasted this for the backport flag.  For consistency, change\nthis to NoOp (a later change will lint this).\n\nAs the prior change, this updates all NoBlock ACLs to use a\nsubmit-requirement of \"is:true\" which should be the same no-op\nsemantics.\n\nChange-Id: I83160aeec0a450f8678ecb583fb7570ac0e71f4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5efd530486af8659e81dd55a8980646ce6d9c9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5efd530486af8659e81dd55a8980646ce6d9c9aa"}]},"branch":"refs/heads/master"},"d21c607eaedbeb713db2d50c7a7109e1253ae43b":{"kind":"REWORK","_number":4,"created":"2023-03-07 04:03:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/875993/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/875993/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/875993/4"}}},"commit":{"parents":[{"commit":"4dd31aac469c7e271bb5d213749e13547cd3549a","subject":"acls : handle submit requirements in normalise tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4dd31aac469c7e271bb5d213749e13547cd3549a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 02:56:42.000000000","tz":660},"subject":"gerrit/acl : convert NoBlock to NoOp","message":"gerrit/acl : convert NoBlock to NoOp\n\nNoBlock is the same as NoOp; for consistency in the ACL files convert\nall to NoOp.  The exact future of the \"function\" in labels and the\nrelationship to submit-requirements remains unclear, but having all\nour ACL\u0027s at least use the same function is one less thing to worry\nabout as we move forward.\n\nThis should have no affect on the rules at all.\n\nNote this does not add a submit-requirement.  These are \"trigger\"\nvotes; i.e. do not have an affect on submit-ability.  An earlier\nrevision did add a submit-requirement with a default of \"is:true\" to\nall these as a future-proof mechanism; but after experimenting on a\nlive server I feel it makes the UI needlessly confusing.  As noted,\nthe function argument is deprecated and it\u0027s unclear if labels will\nhave argument removed in the future, and if they do, how you would\nrepresent these types of labels with submit-requirements.  We will\njust have to cross that bridge when we get to it.\n\nChange-Id: I83160aeec0a450f8678ecb583fb7570ac0e71f4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d21c607eaedbeb713db2d50c7a7109e1253ae43b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d21c607eaedbeb713db2d50c7a7109e1253ae43b"}]},"branch":"refs/heads/master"},"351c09953d7a885a8d39dadd4efc5bc30a556521":{"kind":"REWORK","_number":5,"created":"2023-03-07 22:21:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/875993/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/875993/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/875993/5"}}},"commit":{"parents":[{"commit":"45a00c2f64acfe652f5b574607c7b93b9f2b280e","subject":"gerrit/acl : submit-requirements for deprecated NoOp function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45a00c2f64acfe652f5b574607c7b93b9f2b280e"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 22:21:27.000000000","tz":660},"subject":"gerrit/acl : Convert Backport-Candidate to submit-requirements","message":"gerrit/acl : Convert Backport-Candidate to submit-requirements\n\nSimilar to Ic43f561174ebf30474b1b54be2bed02695cebedc, this converts\nthe existing Backport-Candidate rules to submit-requirements following\nthe gerrit migration rules from [1].\n\nThe function is left as NoBlock, and a no-op submit-requirement is\nadded with the same name.\n\nAs with the prior change; the submit-requirement with the same name as\nthe label will avoid us having migrations run on these ACL\u0027s and keep\nour gerrit in-sync with project-config.\n\nAlso as with the prior change, this should have no affect for users.\n\n[1] https://gerrit-review.googlesource.com/c/gerrit/+/339542\n\nChange-Id: I83160aeec0a450f8678ecb583fb7570ac0e71f4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/351c09953d7a885a8d39dadd4efc5bc30a556521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/351c09953d7a885a8d39dadd4efc5bc30a556521"}]},"branch":"refs/heads/master"},"69c2976cab6cfe860a35e8a63e6cdf0177fe592c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-07 23:13:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/875993/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/875993/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/875993/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/875993/6"}}},"commit":{"parents":[{"commit":"dda3b6098ead3028379afb1ab1b0de9155726b70","subject":"gerrit/acl : submit-requirements for deprecated NoOp function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dda3b6098ead3028379afb1ab1b0de9155726b70"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 22:50:53.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 23:12:23.000000000","tz":660},"subject":"gerrit/acl : Convert Backport-Candidate to submit-requirements","message":"gerrit/acl : Convert Backport-Candidate to submit-requirements\n\nSimilar to Ic43f561174ebf30474b1b54be2bed02695cebedc, this converts\nthe existing Backport-Candidate rules to submit-requirements following\nthe gerrit migration rules from [1].\n\nThe function is left as NoBlock, and a no-op submit-requirement is\nadded with the same name.\n\nAs with the prior change; the submit-requirement with the same name as\nthe label will avoid us having migrations run on these ACL\u0027s and keep\nour gerrit in-sync with project-config.\n\nAlso as with the prior change, this should have no affect for users.\n\n[1] https://gerrit-review.googlesource.com/c/gerrit/+/339542\n\nChange-Id: I83160aeec0a450f8678ecb583fb7570ac0e71f4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69c2976cab6cfe860a35e8a63e6cdf0177fe592c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69c2976cab6cfe860a35e8a63e6cdf0177fe592c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
