)]}'
{"id":"openstack%2Fproject-config~875994","triplet_id":"openstack%2Fproject-config~master~Iada801bd1c38dd1e0502bebefd6a1421c746c90a","project":"openstack/project-config","branch":"master","topic":"gerrit-s-r-3.7","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-03-13 22:41:47.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"Iada801bd1c38dd1e0502bebefd6a1421c746c90a","subject":"gerrit/acl : handle key / values with multiple \u003d","status":"MERGED","created":"2023-03-02 00:18:18.000000000","updated":"2023-03-13 22:48:51.000000000","submitted":"2023-03-13 22:48:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875994-gerrit-s-r-3.7","meta_rev_id":"78ce36bc46fe5795020527e590b10f6b434c26f6","_number":875994,"virtual_id_number":875994,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 22:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-03-07 23:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-03-07 23:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-13 22:41:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 00:45:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 17:21:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-03-07 07:39:50.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"daf4e2835892fdbb4a0a55d9c0b33e50209e89b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 00:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11bfe40647c10186a69896a616011f3851ce47ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 00:45:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e32498bc26e4cbfbec1778d7094820f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c4594afb6d5542278b3d8f9822d9f563 : SUCCESS in 4m 24s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/355db67d3d4d4d9cb46dd2f1c8c8462b : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9437571919cc4c27085ea8a0d36dddf8c529db96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 05:48:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a9145886d398b0f384cca0098ee062d0222bbabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 05:56:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d4abd74d01f4f2ca881160e2e2e845c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cc2d8633a0d84189aac09a68c7ac8ddf : SUCCESS in 4m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bbcfe851a6c34444b6b08dfaeed437bd : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"268a40f4ddc0e1661456c75aca46f4342002ea68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 02:59:49.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":2},{"id":"143e568cb77000cb36809ac41f25f55655db551d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 02:59:51.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":2},{"id":"90289bc6bc4225ffd8b0e74e9aad4b299f07583a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 03:42:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"49e101f2e2a1a4f6b2ea7089dde03119e7253907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:11:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5b7d653f3e48d3b4d0d7e09e565195\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b97b07ae0191405c8760794e58a1e9ce : SUCCESS in 3m 04s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eb5470c6242c4db88f4bb0e9c4a6283c : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8ecb36b07f5c5bd1e4c39dff99cdcd3787d1e015","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-06 17:21:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8630829124db58c12f1ea3de881a8901951bb929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 04:03:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3b7ce4da3044a613a79ba1047ec0bb40301d6b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 04:11:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1304a35d0d847efb4c62cc9a09a0f93\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/571370ec731440e58e3b0478b71a1f8f : SUCCESS in 2m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e5addd47bdfe4542a6fe7efd9bb46fd6 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":4},{"id":"7c8353d75390ca9bb3206929c0f2062da7a0d03d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-03-07 07:39:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bf37f1f33302b439b427ad7bf8eaecb9e7d01cc5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:15:29.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":4},{"id":"2bb0472d2ebb4fcded25f23f6d327d4324149b41","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:15:35.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":4},{"id":"aba15d05ec515b969866c580df7a1baf97facd55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 22:21:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7450aa387c6571dfea8b3d822f651bf95a35d5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:39:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c248420c905494fb80b8f10d58ac8d8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e446db4c700d47d4ae67048d6a33c614 : SUCCESS in 2m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/422bec8c25fa41ada78bcc1bff31c353 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"3d1e98e62c49b2123761e9be7bc858812cc8a91e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 23:13:16.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"1c31263b0c90cd4a89a8ef2a522ceb0f0dffcef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 23:39:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60ef84d71e6146649bb8e6bf44743977\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b1db3a99936a4c30a890ee08ee9e1b24 : SUCCESS in 2m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/78a95fe9c90d4168ac28625fb1b2546c : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":6},{"id":"c4256bbd98eb938c4cb12cd2dd83485cb2148c68","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-13 22:41:47.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1449dacd391cbb03e3a2fd76503fb3efa4c62c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 22:42:07.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"86ef4ebef7c28790c31c8b6f640b97d8f279aabf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 22:48:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4038e1f93ebd40508dc7b98497144a22\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/92ae683f1b9c47e0bab42140cec86de0 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":6},{"id":"78ce36bc46fe5795020527e590b10f6b434c26f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 22:48:51.000000000","message":"Change has been successfully merged\n\n4 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f73a6789456decccd55f173d5e4fab0b487e4f2e","revisions":{"2bee95be3fe5cf233df12f27c8598c35b91400e3":{"kind":"REWORK","_number":1,"created":"2023-03-02 00:18:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/875994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/875994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/875994/1"}}},"commit":{"parents":[{"commit":"2c4b0009742f81ca8024c5a4d948dfa18a2943f8","subject":"gerrit/acl : convert NoBlock to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c4b0009742f81ca8024c5a4d948dfa18a2943f8"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:09:40.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:13:10.000000000","tz":660},"subject":"gerrit/acl : handle key / values with multiple \u003d","message":"gerrit/acl : handle key / values with multiple \u003d\n\nFor things like submit-requirements, we have fields like submittableIf\nwhich take a query string that may have \"\u003d\" on the LHS.  Change the\nkey/value split so that it only takes the key up to the first \"\u003d\".\n\nChange-Id: Iada801bd1c38dd1e0502bebefd6a1421c746c90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bee95be3fe5cf233df12f27c8598c35b91400e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bee95be3fe5cf233df12f27c8598c35b91400e3"}]},"branch":"refs/heads/master"},"5d231ec0b867f900ed844069825d26684fadd32c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-02 05:48:20.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/875994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/875994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/875994/2"}}},"commit":{"parents":[{"commit":"e8b6fc0bb244d3f8c40e58374c2e599abd979be5","subject":"gerrit/acl : add submit requirements to NoBlock labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8b6fc0bb244d3f8c40e58374c2e599abd979be5"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:09:40.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 05:17:54.000000000","tz":660},"subject":"gerrit/acl : handle key / values with multiple \u003d","message":"gerrit/acl : handle key / values with multiple \u003d\n\nFor things like submit-requirements, we have fields like submittableIf\nwhich take a query string that may have \"\u003d\" on the LHS.  Change the\nkey/value split so that it only takes the key up to the first \"\u003d\".\n\nChange-Id: Iada801bd1c38dd1e0502bebefd6a1421c746c90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d231ec0b867f900ed844069825d26684fadd32c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d231ec0b867f900ed844069825d26684fadd32c"}]},"branch":"refs/heads/master"},"4248473f4135f8bdef13e671894f6a72f0e4a362":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-03 03:42:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/875994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/875994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/875994/3"}}},"commit":{"parents":[{"commit":"5efd530486af8659e81dd55a8980646ce6d9c9aa","subject":"gerrit/acl : add submit requirements to NoBlock labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5efd530486af8659e81dd55a8980646ce6d9c9aa"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:09:40.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-03 03:39:25.000000000","tz":660},"subject":"gerrit/acl : handle key / values with multiple \u003d","message":"gerrit/acl : handle key / values with multiple \u003d\n\nFor things like submit-requirements, we have fields like submittableIf\nwhich take a query string that may have \"\u003d\" on the LHS.  Change the\nkey/value split so that it only takes the key up to the first \"\u003d\".\n\nChange-Id: Iada801bd1c38dd1e0502bebefd6a1421c746c90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4248473f4135f8bdef13e671894f6a72f0e4a362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4248473f4135f8bdef13e671894f6a72f0e4a362"}]},"branch":"refs/heads/master"},"c7a0916ad6f0030499e131a4894b1ba82ec00f07":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-07 04:03:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/875994/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/875994/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/875994/4"}}},"commit":{"parents":[{"commit":"d21c607eaedbeb713db2d50c7a7109e1253ae43b","subject":"gerrit/acl : convert NoBlock to NoOp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d21c607eaedbeb713db2d50c7a7109e1253ae43b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:09:40.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 03:04:37.000000000","tz":660},"subject":"gerrit/acl : handle key / values with multiple \u003d","message":"gerrit/acl : handle key / values with multiple \u003d\n\nFor things like submit-requirements, we have fields like submittableIf\nwhich take a query string that may have \"\u003d\" on the LHS.  Change the\nkey/value split so that it only takes the key up to the first \"\u003d\".\n\nChange-Id: Iada801bd1c38dd1e0502bebefd6a1421c746c90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7a0916ad6f0030499e131a4894b1ba82ec00f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7a0916ad6f0030499e131a4894b1ba82ec00f07"}]},"branch":"refs/heads/master"},"06689f91108cadfd1a88ed0786111829ef30d6fc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-07 22:21:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/875994/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/875994/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/875994/5"}}},"commit":{"parents":[{"commit":"351c09953d7a885a8d39dadd4efc5bc30a556521","subject":"gerrit/acl : Convert Backport-Candidate to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/351c09953d7a885a8d39dadd4efc5bc30a556521"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:09:40.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 22:21:28.000000000","tz":660},"subject":"gerrit/acl : handle key / values with multiple \u003d","message":"gerrit/acl : handle key / values with multiple \u003d\n\nFor things like submit-requirements, we have fields like submittableIf\nwhich take a query string that may have \"\u003d\" on the LHS.  Change the\nkey/value split so that it only takes the key up to the first \"\u003d\".\n\nChange-Id: Iada801bd1c38dd1e0502bebefd6a1421c746c90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/06689f91108cadfd1a88ed0786111829ef30d6fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/06689f91108cadfd1a88ed0786111829ef30d6fc"}]},"branch":"refs/heads/master"},"f73a6789456decccd55f173d5e4fab0b487e4f2e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-07 23:13:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/875994/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/875994/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/875994/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/875994/6"}}},"commit":{"parents":[{"commit":"69c2976cab6cfe860a35e8a63e6cdf0177fe592c","subject":"gerrit/acl : Convert Backport-Candidate to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69c2976cab6cfe860a35e8a63e6cdf0177fe592c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:09:40.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 23:12:23.000000000","tz":660},"subject":"gerrit/acl : handle key / values with multiple \u003d","message":"gerrit/acl : handle key / values with multiple \u003d\n\nFor things like submit-requirements, we have fields like submittableIf\nwhich take a query string that may have \"\u003d\" on the LHS.  Change the\nkey/value split so that it only takes the key up to the first \"\u003d\".\n\nChange-Id: Iada801bd1c38dd1e0502bebefd6a1421c746c90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f73a6789456decccd55f173d5e4fab0b487e4f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f73a6789456decccd55f173d5e4fab0b487e4f2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
