)]}'
{"id":"openstack%2Fproject-config~875995","triplet_id":"openstack%2Fproject-config~master~Id5157b9f59082485b6aff92c4f3527fb4c8084aa","project":"openstack/project-config","branch":"master","topic":"gerrit-s-r-3.7","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-03-13 23:48:02.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"Id5157b9f59082485b6aff92c4f3527fb4c8084aa","subject":"gerrit/acl : Convert Review-Priority to submit-requirements","status":"MERGED","created":"2023-03-02 00:18:18.000000000","updated":"2023-03-14 00:35:49.000000000","submitted":"2023-03-13 23:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875995-gerrit-s-r-3.7","meta_rev_id":"2f1c2bb04dbe22ed8eccd3771bd728afc375f22a","_number":875995,"virtual_id_number":875995,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 23:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-03-07 23:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-13 23:48:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 00:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 17:23:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"6433570711584688462b7a802ca98eef8fa967b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 00:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"272b0800d8d78182221a9de4b01844a8cec6d3a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 00:31:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/248df1aa079442fb86c9ab84e2f7b53c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cb220a3a0313445ab0b4ac6127af2c0c : SUCCESS in 3m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ce210765aaff4596a95fad5db70cee55 : SUCCESS in 6m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/95b57a93df4e4ae78213d64199eb6cd1 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1681ec93d0194f757ffb58a79a20bc15174223a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 05:48:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"60d269cd737a938b2062e78c7e84b7a24e82bf7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 05:57:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/379e41a6537d4c0c9c85649116a240dd\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/056e89f2dbb04bb1aaf9f1537fc57e4d : SUCCESS in 3m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9e6c2bd3dff945f488a348651eb9b714 : SUCCESS in 6m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2fb8d9bf8052415197d7dbbe3527f660 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"3ec5d5199973046c0895747e81e28f2b83563f43","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 02:59:43.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":2},{"id":"1851c910b5a00cb128733ed0259d62682e44d226","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 02:59:46.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":2},{"id":"4018ccec80335b6029c5750e3821ad5ccc154b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 03:42:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bbd37f350268fbe4825ef7be4753928bcf0a7f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:22:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/896afb2376a14de9a7c943e8b383b86a\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1e005bce2ecf4ed096a040311494320b : SUCCESS in 4m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a36fcbaaa08441c4a165a8fa27f87065 : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/25b4af03eefe4297ab5ced8a7a1fee3b : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3d254d3f821f81ac323c0cf3ca011de3ab7042c8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-06 17:23:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a86366735ed10dbb3b41eb47a1d3a255de26800e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 04:03:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8eb3d55c9b21d37bff8e0add0b8effdd397a1a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 04:14:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546468c736e34594a2fb023dc35a7e03\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e0c3fcee3bd84cafbd1f4291ae6e26de : SUCCESS in 3m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e52d672a8f654b36a288d6ca0d6fd54a : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3978022e49fd44cf8663a708df478a7d : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":4},{"id":"cd4aae58fba672e2a0db27aedbeb822f441b095e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:15:39.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":4},{"id":"21c4e024402062be3e2b1cae422352b71e6008ee","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:15:46.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":4},{"id":"72ae6fb21e8c8599d7021a1619c928bc38636fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 22:21:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a910707c42a63ba2da72ba1050ad6c10dd2f4a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:28:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d9d11d84124ca791ebeeb86a38a585\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f2a71157f61244bb87f3ca6ce33d0617 : SUCCESS in 2m 45s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/752a514dbf72459baea6a525064df0ed : SUCCESS in 5m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e6741f940c944e2faf99098a70c4e042 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":5},{"id":"dc2eff55cfc357c6be14090671ba14a8de0c6881","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-07 23:10:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bd00c966a1e29c9bcca387237ca8a5b90493fc88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 23:13:16.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c40cb928509ecdc04fa92ef61784dc5e7d92b2de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 23:22:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d03beb8677d4fc08a1bc9f9f0dc247b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d152cdf55eb9434f8cbe5b69a3b4f936 : SUCCESS in 2m 59s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/96094699a53147ddbbb5cffbda9442aa : SUCCESS in 5m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/34b53cacc8b042e180ad80972fb1a868 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":6},{"id":"6adf5d4ce6909a133c43af678c17120313eaa5d3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-13 23:48:02.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"59420aff0207340597d42839e859e6b35b2a093f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 23:48:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"dd6c26d1736275e613e1fb939839dcc2f6a62a08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 23:54:33.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d0f78ac0c8c4464852b27b78134c9db\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a90d2cdc9d7641beb88205f814ba03d6 : SUCCESS in 6m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d88d9afa063141bc82d1b868357996c9 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"6351dd958a01fc58efc7d6136bc738cef16f4ca2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 23:54:33.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"2f1c2bb04dbe22ed8eccd3771bd728afc375f22a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 00:35:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8603b1df3914add9ceb66381a1e7053\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/51cdfcf6737b4438aea5480020214022 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"861d3fb4480b30fe5bd64afdb7b62477a3f9842f","revisions":{"5f160817acc652bb5b588d671eb98065925ce8c3":{"kind":"REWORK","_number":1,"created":"2023-03-02 00:18:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/875995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875995/1"}}},"commit":{"parents":[{"commit":"2bee95be3fe5cf233df12f27c8598c35b91400e3","subject":"gerrit/acl : handle key / values with multiple \u003d","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bee95be3fe5cf233df12f27c8598c35b91400e3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:20:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:13:10.000000000","tz":660},"subject":"gerrit/acl : Update Review-Priority to submit-requirements","message":"gerrit/acl : Update Review-Priority to submit-requirements\n\nThis updates Review-Priority labels to use submit requirements.  These\nare all using \"AnyWithBlock\" which means that only the lowest possible\nvote blocks submission.  Thus we replace this with a\nsubmit-requirement of \"-label:Review-Priority\u003dMIN\".\n\nChange-Id: Id5157b9f59082485b6aff92c4f3527fb4c8084aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f160817acc652bb5b588d671eb98065925ce8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f160817acc652bb5b588d671eb98065925ce8c3"}]},"branch":"refs/heads/master"},"15ba794e2072f69f9146810e6ee13e27da0e9ce0":{"kind":"REWORK","_number":2,"created":"2023-03-02 05:48:20.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/875995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875995/2"}}},"commit":{"parents":[{"commit":"5d231ec0b867f900ed844069825d26684fadd32c","subject":"gerrit/acl : handle key / values with multiple \u003d","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d231ec0b867f900ed844069825d26684fadd32c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:20:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 05:25:38.000000000","tz":660},"subject":"gerrit/acl : Update Review-Priority to submit-requirements","message":"gerrit/acl : Update Review-Priority to submit-requirements\n\nThis updates Review-Priority labels to use submit requirements.  These\nare all using \"AnyWithBlock\" which means that only the lowest possible\nvote blocks submission.  Thus we replace this with a\nsubmit-requirement of \"-label:Review-Priority\u003dMIN\".\n\nAs the submit-requirements transition is happening, You currently can\nnot delete the \"function\" from a label.  Thus we make it NoOp here.\n\nChange-Id: Id5157b9f59082485b6aff92c4f3527fb4c8084aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15ba794e2072f69f9146810e6ee13e27da0e9ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15ba794e2072f69f9146810e6ee13e27da0e9ce0"}]},"branch":"refs/heads/master"},"7e980a8a945cecc685dc1fee7c58933e0309d865":{"kind":"REWORK","_number":3,"created":"2023-03-03 03:42:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/875995/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875995/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875995/3"}}},"commit":{"parents":[{"commit":"4248473f4135f8bdef13e671894f6a72f0e4a362","subject":"gerrit/acl : handle key / values with multiple \u003d","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4248473f4135f8bdef13e671894f6a72f0e4a362"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:20:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-03 03:40:34.000000000","tz":660},"subject":"gerrit/acl : Update Review-Priority to submit-requirements","message":"gerrit/acl : Update Review-Priority to submit-requirements\n\nThis updates Review-Priority labels to use submit requirements.  These\nare all using \"AnyWithBlock\" which means that only the lowest possible\nvote blocks submission.  Thus we replace this with a\nsubmit-requirement of \"-label:Review-Priority\u003dMIN\".\n\nAs the submit-requirements transition is happening, You currently can\nnot delete the \"function\" from a label.  Thus we make it NoOp here.\n\nChange-Id: Id5157b9f59082485b6aff92c4f3527fb4c8084aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e980a8a945cecc685dc1fee7c58933e0309d865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e980a8a945cecc685dc1fee7c58933e0309d865"}]},"branch":"refs/heads/master"},"1e596878d513c5c03667f747e62fa8daa81f6b17":{"kind":"REWORK","_number":4,"created":"2023-03-07 04:03:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/875995/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875995/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875995/4"}}},"commit":{"parents":[{"commit":"c7a0916ad6f0030499e131a4894b1ba82ec00f07","subject":"gerrit/acl : handle key / values with multiple \u003d","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7a0916ad6f0030499e131a4894b1ba82ec00f07"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:20:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 03:12:30.000000000","tz":660},"subject":"gerrit/acl : Update Review-Priority to submit-requirements","message":"gerrit/acl : Update Review-Priority to submit-requirements\n\nThis updates Review-Priority labels to use submit requirements.  These\nare all using \"AnyWithBlock\" which means that only the lowest possible\nvote blocks submission.  Thus we replace this with a\nsubmit-requirement of \"-label:Review-Priority\u003dMIN\".\n\nAs the submit-requirements transition is happening, You currently can\nnot delete the \"function\" from a label.  Thus we make it NoOp here.\n\nChange-Id: Id5157b9f59082485b6aff92c4f3527fb4c8084aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e596878d513c5c03667f747e62fa8daa81f6b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e596878d513c5c03667f747e62fa8daa81f6b17"}]},"branch":"refs/heads/master"},"54e0ae7021956c424c9032426ae9b992156b5182":{"kind":"REWORK","_number":5,"created":"2023-03-07 22:21:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/875995/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875995/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875995/5"}}},"commit":{"parents":[{"commit":"06689f91108cadfd1a88ed0786111829ef30d6fc","subject":"gerrit/acl : handle key / values with multiple \u003d","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/06689f91108cadfd1a88ed0786111829ef30d6fc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:20:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 22:21:28.000000000","tz":660},"subject":"gerrit/acl : Convert Review-Priority to submit-requirements","message":"gerrit/acl : Convert Review-Priority to submit-requirements\n\nSimilar to Ic43f561174ebf30474b1b54be2bed02695cebedc and\nI83160aeec0a450f8678ecb583fb7570ac0e71f4a, this converts the existing\nReview-Priority rules to submit-requirements following the gerrit\nmigration rules from [1].\n\nThese are all using \"AnyWithBlock\" which means that only the lowest\npossible vote blocks submission.  Thus we replace this with a\nsubmit-requirement of \"-label:Review-Priority\u003dMIN\".  The function is\nchanged to NoBlock as done by the migration tool.\n\nAs with the prior change; the submit-requirement with the same name as\nthe label will avoid us having migrations run on these ACL\u0027s and\nkeep our gerrit in-sync with project-config.\n\nAlso as with the prior change, this should have no affect for users.\n\n[1] https://gerrit-review.googlesource.com/c/gerrit/+/339542\n\nChange-Id: Id5157b9f59082485b6aff92c4f3527fb4c8084aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/54e0ae7021956c424c9032426ae9b992156b5182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/54e0ae7021956c424c9032426ae9b992156b5182"}]},"branch":"refs/heads/master"},"861d3fb4480b30fe5bd64afdb7b62477a3f9842f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-07 23:13:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/95/875995/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/875995/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/875995/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/875995/6"}}},"commit":{"parents":[{"commit":"f73a6789456decccd55f173d5e4fab0b487e4f2e","subject":"gerrit/acl : handle key / values with multiple \u003d","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f73a6789456decccd55f173d5e4fab0b487e4f2e"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:20:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 23:12:23.000000000","tz":660},"subject":"gerrit/acl : Convert Review-Priority to submit-requirements","message":"gerrit/acl : Convert Review-Priority to submit-requirements\n\nSimilar to Ic43f561174ebf30474b1b54be2bed02695cebedc and\nI83160aeec0a450f8678ecb583fb7570ac0e71f4a, this converts the existing\nReview-Priority rules to submit-requirements following the gerrit\nmigration rules from [1].\n\nThese are all using \"AnyWithBlock\" which means that only the lowest\npossible vote blocks submission.  Thus we replace this with a\nsubmit-requirement of \"-label:Review-Priority\u003dMIN\".  The function is\nchanged to NoBlock as done by the migration tool.\n\nAs with the prior change; the submit-requirement with the same name as\nthe label will avoid us having migrations run on these ACL\u0027s and\nkeep our gerrit in-sync with project-config.\n\nAlso as with the prior change, this should have no affect for users.\n\n[1] https://gerrit-review.googlesource.com/c/gerrit/+/339542\n\nChange-Id: Id5157b9f59082485b6aff92c4f3527fb4c8084aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/861d3fb4480b30fe5bd64afdb7b62477a3f9842f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/861d3fb4480b30fe5bd64afdb7b62477a3f9842f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
