)]}'
{"id":"openstack%2Fproject-config~875996","triplet_id":"openstack%2Fproject-config~master~Id57bdb4f6ed38a90a37a738068446197e6a24769","project":"openstack/project-config","branch":"master","topic":"gerrit-s-r-3.7","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-03-14 00:23:10.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-03-14 05:32:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"Id57bdb4f6ed38a90a37a738068446197e6a24769","subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","status":"MERGED","created":"2023-03-02 00:18:18.000000000","updated":"2023-03-14 06:21:56.000000000","submitted":"2023-03-14 05:49:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875996-gerrit-s-r-3.7","meta_rev_id":"18c1ef1ab8f2db42190c2900a20efa9890dbf9d8","_number":875996,"virtual_id_number":875996,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-14 05:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-03-14 00:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-14 05:32:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 00:36:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 17:24:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"5e55402b812a626843c0be198b99c2a1adc1e96e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 00:18:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce432a42154a2682a42e2ec7c0231ce306258388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 00:36:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dbaae83c26a4119b2cabedf6ce1b528\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5f00a78ad4ab40b09ee449dd753affa6 : SUCCESS in 2m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5c2c3207a42445d492260594ce1e4a4e : SUCCESS in 5m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a4c91650916f49769a3db85ca99dddb6 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c126acec7a39d39a1599c73b2218d47263c99edc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-02 05:48:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"aa2a436b5fc859841a6bae95cac286616010aecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 05:57:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/475cbe7ed7b34581b67550c1eb1c90ce\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/26c31c4bf5bb475b8e0ad843f6b3ac4a : SUCCESS in 4m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4195861ffdf4193a5e9847ee775634e : SUCCESS in 6m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9c957851ae014f3eaa99253b0f5fd95a : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"9b522b901c4d3429feac37a4a4ab5d0af83e2a0a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 02:59:37.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":2},{"id":"38a3db9a6099dfb772aafb47fe5c71632d8e0a83","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 02:59:40.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":2},{"id":"31c476691a0fd1e77d1cee871abbbc8ca1d076e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-03 03:42:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"483266907db7035a43c004347676a5c46dcb55b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:26:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d6ecc4efeb46059c251a0db4498abc\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/11236ae0fca540c699ecc55bec0b000f : SUCCESS in 3m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dfb8d45c7e494566848b08d38f9f14c2 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/36d1509ba6c3422b8de05ece0e6b725e : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"b516fef41fe7a68baa1fb434e180b0b1bf822f8f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-06 17:24:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a680adf550833d654f8ce03cd21e74fa8725e7f3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 04:01:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f28427fb15d5d27f21a80d27bc7dd8042d48f1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 04:03:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cfff88de38bfffc5f72ee39387bb10bd349dcc43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 04:12:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e40ce228fe3448cabe2893019123bcb2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b9ecd78bcffd4e3ab1a4519775c709f3 : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9878e5f5c9e243e8820e7130a66b3d93 : SUCCESS in 5m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6a539a01623f4e819454aedaafa16c8f : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"116d44e44db9a92c12001cceb511df36c9a81dea","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:15:48.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":4},{"id":"e64512de30ca346e9f3f66ec8935a5d377833576","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-07 19:15:53.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":4},{"id":"dad74a33c6362cc137c8ab2b696d5cf222b65fa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-14 00:12:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0dbdd95a3ba51a16cc612602c7891da54be2124b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-14 00:12:38.000000000","message":"Topic copyCondition removed","accounts_in_message":[],"_revision_number":5},{"id":"96abf5b1eeab9bcbe4e44fb034d2d22ce4478e2d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-14 00:12:45.000000000","message":"Topic set to gerrit-s-r-3.7","accounts_in_message":[],"_revision_number":5},{"id":"320f80db17f839ec227cfdba5c44775e59ed429f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-14 00:21:01.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9b1188355774f622e0cf48c27d31f27832c7d936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-14 00:22:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"686b3f7db8d23cf89a6efc6d023405cbb3b64d36","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-14 00:22:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5fc21f1059cf4aa0be97822583527924a5974c4a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-14 00:23:10.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"be68d72858aa7551b6700351a23dd37870928b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 00:33:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c850abced3f4cc7be859a5237749442\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6ab99bd68f264376a47b1a3b4868c390 : SUCCESS in 2m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/039165e512d343ada4a58cf07ae08619 : SUCCESS in 6m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cad453070f404d42bde0bd2a6e3da3a5 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":6},{"id":"20bfb98fbfbd959f33a1911ca2f7c9e92369ae39","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-14 05:32:10.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6060f56a59866fa84afa502a03fbe03d97f6765c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 05:32:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"747f237ebffca257d49472266e9aed97a2a2f15b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 05:49:52.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18891640cab4820b0b5cdf3d5052c6d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e06afe9a69c49e0a09912347b962310 : SUCCESS in 5m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eb503535b3c447a3a1ca0989e60bff8a : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":6},{"id":"291269ac97da75781641fe1280750bae24814213","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 05:49:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"18c1ef1ab8f2db42190c2900a20efa9890dbf9d8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 06:21:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c895b893c58c4a6689e58d4ec15d1e68\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/7e71165c83d24cbe9073bc58e977b2fb : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1c0400db533960ad252dedba4c2676745192a262","revisions":{"411fef7c2932334a27c6f71a164b6f338c079220":{"kind":"REWORK","_number":1,"created":"2023-03-02 00:18:18.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/875996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/875996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/875996/1"}}},"commit":{"parents":[{"commit":"5f160817acc652bb5b588d671eb98065925ce8c3","subject":"gerrit/acl : Update Review-Priority to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f160817acc652bb5b588d671eb98065925ce8c3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:33:11.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 00:13:10.000000000","tz":660},"subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","message":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements\n\nPrior change Id5157b9f59082485b6aff92c4f3527fb4c8084aa updated the\nReview-Priority label; this converts the couple of remaining\nAnyWithBlock labels that didn\u0027t match the review-priority pattern to\nsubmit requirements.\n\nChange-Id: Id57bdb4f6ed38a90a37a738068446197e6a24769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/411fef7c2932334a27c6f71a164b6f338c079220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/411fef7c2932334a27c6f71a164b6f338c079220"}]},"branch":"refs/heads/master"},"2d5ad87ff5bb3d380f6e29cb6ba11fe22e1beeaa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-02 05:48:20.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/875996/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/875996/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/875996/2"}}},"commit":{"parents":[{"commit":"15ba794e2072f69f9146810e6ee13e27da0e9ce0","subject":"gerrit/acl : Update Review-Priority to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15ba794e2072f69f9146810e6ee13e27da0e9ce0"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:33:11.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-02 05:27:14.000000000","tz":660},"subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","message":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements\n\nPrior change Id5157b9f59082485b6aff92c4f3527fb4c8084aa updated the\nReview-Priority label; this converts the couple of remaining\nAnyWithBlock labels that didn\u0027t match the review-priority pattern to\nsubmit requirements.\n\nChange-Id: Id57bdb4f6ed38a90a37a738068446197e6a24769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d5ad87ff5bb3d380f6e29cb6ba11fe22e1beeaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d5ad87ff5bb3d380f6e29cb6ba11fe22e1beeaa"}]},"branch":"refs/heads/master"},"409318a81dd25f65ba1a76044843fb4773b75dbe":{"kind":"REWORK","_number":3,"created":"2023-03-03 03:42:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/875996/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/875996/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/875996/3"}}},"commit":{"parents":[{"commit":"7e980a8a945cecc685dc1fee7c58933e0309d865","subject":"gerrit/acl : Update Review-Priority to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e980a8a945cecc685dc1fee7c58933e0309d865"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:33:11.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-03 03:41:33.000000000","tz":660},"subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","message":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements\n\nPrior change Id5157b9f59082485b6aff92c4f3527fb4c8084aa updated the\nReview-Priority label; this converts the couple of remaining\nAnyWithBlock labels that didn\u0027t match the review-priority pattern to\nsubmit requirements.\n\nChange-Id: Id57bdb4f6ed38a90a37a738068446197e6a24769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/409318a81dd25f65ba1a76044843fb4773b75dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/409318a81dd25f65ba1a76044843fb4773b75dbe"}]},"branch":"refs/heads/master"},"b706cb6538da4fe103d860b9682cde99ab43f783":{"kind":"REWORK","_number":4,"created":"2023-03-07 04:03:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/875996/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/875996/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/875996/4"}}},"commit":{"parents":[{"commit":"1e596878d513c5c03667f747e62fa8daa81f6b17","subject":"gerrit/acl : Update Review-Priority to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e596878d513c5c03667f747e62fa8daa81f6b17"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:33:11.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-07 03:14:49.000000000","tz":660},"subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","message":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements\n\nPrior change Id5157b9f59082485b6aff92c4f3527fb4c8084aa updated the\nReview-Priority label; this converts the couple of remaining\nAnyWithBlock labels that didn\u0027t match the review-priority pattern to\nsubmit requirements.\n\nChange-Id: Id57bdb4f6ed38a90a37a738068446197e6a24769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b706cb6538da4fe103d860b9682cde99ab43f783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b706cb6538da4fe103d860b9682cde99ab43f783"}]},"branch":"refs/heads/master"},"d0632b926dcd2df66db0d163d067df0ddc15e9d5":{"kind":"REWORK","_number":5,"created":"2023-03-14 00:12:22.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/875996/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/875996/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/875996/5"}}},"commit":{"parents":[{"commit":"861d3fb4480b30fe5bd64afdb7b62477a3f9842f","subject":"gerrit/acl : Convert Review-Priority to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/861d3fb4480b30fe5bd64afdb7b62477a3f9842f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:33:11.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-14 00:10:46.000000000","tz":660},"subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","message":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements\n\nPrior change Id5157b9f59082485b6aff92c4f3527fb4c8084aa updated the\nReview-Priority label; this converts the couple of remaining\nAnyWithBlock labels that didn\u0027t match the review-priority pattern to\nsubmit requirements.\n\nChange-Id: Id57bdb4f6ed38a90a37a738068446197e6a24769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0632b926dcd2df66db0d163d067df0ddc15e9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d0632b926dcd2df66db0d163d067df0ddc15e9d5"}]},"branch":"refs/heads/master"},"1c0400db533960ad252dedba4c2676745192a262":{"kind":"REWORK","_number":6,"created":"2023-03-14 00:22:23.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/96/875996/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/875996/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/875996/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/875996/6"}}},"commit":{"parents":[{"commit":"861d3fb4480b30fe5bd64afdb7b62477a3f9842f","subject":"gerrit/acl : Convert Review-Priority to submit-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/861d3fb4480b30fe5bd64afdb7b62477a3f9842f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-01 23:33:11.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-14 00:22:05.000000000","tz":660},"subject":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements","message":"gerrit/acl : Convert remaining AnyWithBlock to submit requirements\n\nPrior change Id5157b9f59082485b6aff92c4f3527fb4c8084aa updated the\nReview-Priority label; this converts the couple of remaining\nAnyWithBlock labels that didn\u0027t match the review-priority pattern to\nsubmit requirements.\n\nChange-Id: Id57bdb4f6ed38a90a37a738068446197e6a24769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c0400db533960ad252dedba4c2676745192a262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1c0400db533960ad252dedba4c2676745192a262"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
