)]}'
{"id":"openstack%2Fproject-config~877569","triplet_id":"openstack%2Fproject-config~master~I557f3615d15eca899a262b0989986fb2754ac870","project":"openstack/project-config","branch":"master","topic":"copyCondition","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-03-16 23:07:39.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I557f3615d15eca899a262b0989986fb2754ac870","subject":"gerrit/acl : fix some missed NoOp functions","status":"MERGED","created":"2023-03-16 01:58:55.000000000","updated":"2023-03-16 23:43:14.000000000","submitted":"2023-03-16 23:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877569-copyCondition","meta_rev_id":"90a081f41cf30fc1d0ebe3c8f20c357be103aa82","_number":877569,"virtual_id_number":877569,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 23:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-03-16 18:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-03-16 07:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-16 23:07:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 02:14:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 07:56:06.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2023-03-16 18:02:59.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"19c95f54f2a7910237328f1cda79288cd86996db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-16 01:58:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee986bc4c74737a33f1459a63d0066c965fabb7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 02:14:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ec1cede04c4d54b4523e2715c5be05\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/96af64edfa6d4e6899160d713f850d94 : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2fc2a6e146bf47c9ac8daa4bcd1b4bb4 : SUCCESS in 7m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dbc8273a306a4af5878705d28456cda0 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5a1af11823473bc8470ac62ba85b0b40b6258c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-16 02:38:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"46216243c84fe4b4bb68d223f849a5ff425207d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 02:49:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5b6e48cc1a41499e4c0b53a69e3040\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e97ed6d406e84d8f8cb14f9bf41f3974 : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/831d553898ad4d24bc4b9f68e33f3121 : SUCCESS in 6m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e068c233035d47a88f7bf3332560bff6 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a6606d9e93c74126f05804fc9fc9b5c34d5b321a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-03-16 07:56:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e57cae50bcefe789a10a29fa0468d49c621603a0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-03-16 18:02:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c9789d8b50bc714eb2f0eb32b6989c5b491c0855","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-03-16 23:07:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4ca32209f76a715bfd10b2623133a1ed5283b98d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 23:07:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b30886f91f08a9ce21356c080f343cf569a9335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 23:20:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/151ab16c58b7468f8636e61a68db365d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c1864426ad54d0e9e0ad0c69a3d2b80 : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3ebac47c15434f1da864d9997b1ee24c : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"9bbff5156b467ded624ac6d540f4112a37e0e783","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 23:20:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"90a081f41cf30fc1d0ebe3c8f20c357be103aa82","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 23:43:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4e81d66d954791b83ee01636ba7675\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/8231216cda374696b4d334707a5c01a3 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2182282f60f04f12c625beb2e9d46b0abd13c5fa","revisions":{"ab7f4a2c6b15e503b4961e59e3360c823357859c":{"kind":"REWORK","_number":1,"created":"2023-03-16 01:58:55.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/69/877569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/877569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/877569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/877569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/877569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/877569/1"}}},"commit":{"parents":[{"commit":"fd2b1e2f2f134d3ae4ad6c4cb4606ce023232fe9","subject":"Add post-review pipeline reporting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd2b1e2f2f134d3ae4ad6c4cb4606ce023232fe9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-16 01:54:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-16 01:55:47.000000000","tz":660},"subject":"gerrit/acls : fix some missed NoOp functions","message":"gerrit/acls : fix some missed NoOp functions\n\nIn prior changes moving to submit-requirements, we standardised on\n\"NoBlock\" instead of \"NoOp\" because that\u0027s what the upstream migration\ntools do.  But it seems we missed some, correct that here.\n\nThis was found by the linter in\nI971f626bd7dbee012dc93a5807145d206b645cfd\n\nChange-Id: I557f3615d15eca899a262b0989986fb2754ac870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab7f4a2c6b15e503b4961e59e3360c823357859c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab7f4a2c6b15e503b4961e59e3360c823357859c"}]},"branch":"refs/heads/master"},"2182282f60f04f12c625beb2e9d46b0abd13c5fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-16 02:38:11.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/69/877569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/877569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/877569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/877569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/877569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/877569/2"}}},"commit":{"parents":[{"commit":"fd2b1e2f2f134d3ae4ad6c4cb4606ce023232fe9","subject":"Add post-review pipeline reporting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd2b1e2f2f134d3ae4ad6c4cb4606ce023232fe9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-16 01:54:31.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-03-16 02:37:45.000000000","tz":660},"subject":"gerrit/acl : fix some missed NoOp functions","message":"gerrit/acl : fix some missed NoOp functions\n\nIn prior changes moving to submit-requirements, we standardised on\n\"NoBlock\" instead of \"NoOp\" because that\u0027s what the upstream migration\ntools do.  But it seems we missed some, correct that here.\n\nThis was found by the linter in\nI971f626bd7dbee012dc93a5807145d206b645cfd\n\nChange-Id: I557f3615d15eca899a262b0989986fb2754ac870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2182282f60f04f12c625beb2e9d46b0abd13c5fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2182282f60f04f12c625beb2e9d46b0abd13c5fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
