)]}'
{"id":"openstack%2Fproject-config~878932","triplet_id":"openstack%2Fproject-config~master~I39c6a1764c6078958058bfed33d436aa9990532e","project":"openstack/project-config","branch":"master","topic":"placement_off_storyboard","attention_set":{},"removed_from_attention_set":{"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-03-31 11:44:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I39c6a1764c6078958058bfed33d436aa9990532e","subject":"Stop using Storyboard for Placement projects","status":"MERGED","created":"2023-03-29 17:53:13.000000000","updated":"2023-04-24 08:25:23.000000000","submitted":"2023-03-31 11:44:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878932-placement_off_storyboard","meta_rev_id":"1a81882060eadfa954fb0589c1e8b7272df89d56","_number":878932,"virtual_id_number":878932,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 11:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-03-30 14:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":2,"date":"2023-03-30 14:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-03-31 11:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 18:01:04.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-03-29 18:05:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 14:25:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-03-30 14:38:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6f32281adf44b9e1adb939c6f308ccd1317b8159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-29 17:53:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d601559ab90f2a53f227d40e0bd08c488b56a32","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-29 18:01:04.000000000","message":"Patch Set 1:\n\nNote that by leaving the \"groups: placement\" in those entries, bug update comments from gerrit will assume that os-resource-classes, os-traits, and osc-placement are tracking their bugs in https://launchpad.net/placement so make sure that\u0027s what you intend.","accounts_in_message":[],"_revision_number":1},{"id":"d63ede4eca067c88b433ca424c3561a00c600d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 18:05:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52b272dbdd7d4b7a888de52cfa42b0fa\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f65f156729ed4f919bef4fcddca15877 : SUCCESS in 2m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d99b3cccbfa04be485178e2eeea57a6a : SUCCESS in 6m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b2c483387c2b410fb3ed02f2ce31f4c5 : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"30c74d626882fcbb2012932c9fc35b04186eaece","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-29 18:20:10.000000000","message":"Patch Set 1:\n\nAlso the linter failure is because the placement project also sets \"groups: placement\" which is redundant when using Launchpad.","accounts_in_message":[],"_revision_number":1},{"id":"507c2da7b7a884e392b50c9ea5c10d5f5707633d","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-30 07:35:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecb03f33ea3a88f6f5425c59d63649d6626aca2a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-30 07:39:40.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cf492fc7fd241c4b0561609624f4120e600e879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-03-30 07:39:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c427d6f2da7c8058345a46ff3aeb0e8b285da78a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 07:48:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05044b1934524ddfa971b44518feea87\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ecfbfe50d33f4b87a378f623fefcc63b : SUCCESS in 3m 04s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a7ec8c15507e4191826de2c43d22fd64 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2f67421e2016427db813d2c20e37f202 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c80a61f829a8c0a312ece3c87cbda6023c33394b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-30 14:25:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5c36f2d271b93a005ef64f1a6368bbe97906cf11","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-30 14:38:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2764d4f0005e84a4a00ae510e37c86ae06a1d44","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-30 14:57:51.000000000","message":"Patch Set 2:\n\nIn the past when projects have moved off SB, I\u0027ve set them to \"inactive\" (which doesn\u0027t hide the old stories/tasks but does disable selecting them for adding new tasks I think), and adjusting the project description to indicate where new bugs should be filed.","accounts_in_message":[],"_revision_number":2},{"id":"51d129b48d2093b6b898dc25221d63b0d1fdb05e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-31 11:36:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a35be01c249eca8ad546ad8cbd9a42c5cc10de35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:36:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e56a8aaa3942b80425f2a9c4839066e69fb85956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:44:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544711f326094f429625e55f19d8cfd9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5cb83bec28824492a938065ccb0c0392 : SUCCESS in 6m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0a1e3b7b2eb54c1793c9c1256604ed24 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"56ea79bf4afcf2f20c7815a60fe130de9fd0decb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:44:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d7db6c69050dc7b55ceb2da6e26dd4688f23568e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:59:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/711bfbf9e7364e07921294d057d44e4d\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/46e535d0e95e487a9ce42e9490cbc331 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3fad9d961fac2ec4acf349a6c18cb6598925598f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-21 17:02:53.000000000","message":"Patch Set 2:\n\nI\u0027ve deactivated and updated descriptions in SB for:\n\n  openstack/os-resource-classes\n  openstack/os-traits\n  openstack/osc-placement\n  openstack/placement","accounts_in_message":[],"_revision_number":2},{"id":"1a81882060eadfa954fb0589c1e8b7272df89d56","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-24 08:25:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88e24cbfceeb2dd628d19a30473ff1b500ad90ca","revisions":{"ae23040b3cb3f40b566a33472f91827c3ce005a6":{"kind":"REWORK","_number":1,"created":"2023-03-29 17:53:13.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/32/878932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/878932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/878932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/878932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/878932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/878932/1"}}},"commit":{"parents":[{"commit":"9660ec9f1962ddfcd703485d8ebd7233e6043230","subject":"Merge \"Add metal3-io/metal3-dev-env to openstack tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9660ec9f1962ddfcd703485d8ebd7233e6043230"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-29 17:52:26.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-29 17:52:26.000000000","tz":120},"subject":"Stop using Storyboard for Placement projects","message":"Stop using Storyboard for Placement projects\n\nThe nova community agreed on previous PTG to stop using Storyboard for\ntracking bugs and features.\n\nChange-Id: I39c6a1764c6078958058bfed33d436aa9990532e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae23040b3cb3f40b566a33472f91827c3ce005a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae23040b3cb3f40b566a33472f91827c3ce005a6"}]},"branch":"refs/heads/master"},"88e24cbfceeb2dd628d19a30473ff1b500ad90ca":{"kind":"REWORK","_number":2,"created":"2023-03-30 07:39:49.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/32/878932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/878932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/878932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/878932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/878932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/878932/2"}}},"commit":{"parents":[{"commit":"9660ec9f1962ddfcd703485d8ebd7233e6043230","subject":"Merge \"Add metal3-io/metal3-dev-env to openstack tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9660ec9f1962ddfcd703485d8ebd7233e6043230"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-29 17:52:26.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-03-30 07:37:56.000000000","tz":120},"subject":"Stop using Storyboard for Placement projects","message":"Stop using Storyboard for Placement projects\n\nThe nova community agreed on previous PTG to stop using Storyboard for\ntracking bugs and features.\nAll the placement repositories (osc-placement, placement, os-traits\nand os-resource-classes) will use the same \u0027placement\u0027 Launchpad project.\n\nChange-Id: I39c6a1764c6078958058bfed33d436aa9990532e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88e24cbfceeb2dd628d19a30473ff1b500ad90ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88e24cbfceeb2dd628d19a30473ff1b500ad90ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
