)]}'
{"id":"openstack%2Fproject-config~879196","triplet_id":"openstack%2Fproject-config~master~I27766140120fb55a2eab35552f0321b1da9c67ff","project":"openstack/project-config","branch":"master","topic":"nested-virt-debian","hashtags":[],"change_id":"I27766140120fb55a2eab35552f0321b1da9c67ff","subject":"Add nested-virt-debian-bullseye label to nodepool","status":"MERGED","created":"2023-03-31 16:20:09.000000000","updated":"2023-04-13 06:56:06.000000000","submitted":"2023-04-13 05:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879196-nested-virt-debian","meta_rev_id":"05beab5df20e0833b1b19c58c4f5c70d2673230e","_number":879196,"virtual_id_number":879196,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 05:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 05:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-13 05:37:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 16:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 05:37:02.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"d6388c5d297691f836fa49b369aad9d6e2cbd9e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-31 16:20:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97d2039af6d3a9972c8dffa48b05fdb2efd9ed05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 16:30:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679f8cf6a86642febedb5b5bd4c656ff\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5a05f84e1f554aa6bb6be2a8af6a6e74 : SUCCESS in 2m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1dc7861e986842e3aa297d96cb492197 : SUCCESS in 5m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/39dfb6611a684b64ab509413dbea1fce : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c7150b5202402742a8af3cd4f1762ab701ca2397","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-13 05:37:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba26b6db330ea310ac92e2c5625c0f3d2ff99785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 05:37:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a62c6ae8c0c82dd749446824841841dddc59236e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 05:49:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39321224179b4b6e90cb805503830c07\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fab65366cfd54e9b9d19a4f77871996f : SUCCESS in 6m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3fe7457b65e548a9a394115b4e5a5ce9 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e7e6ef8dd6da2df682e4881fdf8f3fd7563388af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 05:49:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05beab5df20e0833b1b19c58c4f5c70d2673230e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 06:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34698a3d5a0242f2a623a8ccf2dbd502\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/1995b451763742e5b16d87d0779f5734 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac5c9ccc5bbfde379d483c4c6d9a40fcd6e906cf","revisions":{"ac5c9ccc5bbfde379d483c4c6d9a40fcd6e906cf":{"kind":"REWORK","_number":1,"created":"2023-03-31 16:20:09.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/96/879196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/879196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/879196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/879196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/879196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/879196/1"}}},"commit":{"parents":[{"commit":"b0e407064944d165f7a458dfc40671f0286f3ade","subject":"Merge \"Replace old Antelope cycle key with 2023.2/Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b0e407064944d165f7a458dfc40671f0286f3ade"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-31 16:15:25.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-31 16:15:25.000000000","tz":120},"subject":"Add nested-virt-debian-bullseye label to nodepool","message":"Add nested-virt-debian-bullseye label to nodepool\n\nkolla wants to have testing parity between Ubuntu and Debian, so add a\nnested-virt-debian-bullseye label to nodepool matching the existing\nnested-virt-ubuntu-bionic label.\n\nChange-Id: I27766140120fb55a2eab35552f0321b1da9c67ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac5c9ccc5bbfde379d483c4c6d9a40fcd6e906cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac5c9ccc5bbfde379d483c4c6d9a40fcd6e906cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
