)]}'
{"id":"openstack%2Fproject-config~879213","triplet_id":"openstack%2Fproject-config~master~I8ba595466ebc9e0546680be0c16454d9e7903d2c","project":"openstack/project-config","branch":"master","topic":"project-yaml-normalization","hashtags":[],"change_id":"I8ba595466ebc9e0546680be0c16454d9e7903d2c","subject":"Normalize projects.yaml","status":"MERGED","created":"2023-04-01 02:24:24.000000000","updated":"2023-04-01 03:22:11.000000000","submitted":"2023-04-01 02:56:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879213-project-yaml-normalization","meta_rev_id":"47a43e5ba86247b39989e87bda33f1b9d5a101f5","_number":879213,"virtual_id_number":879213,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-01 02:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-01 02:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-01 02:48:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-01 02:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-01 02:48:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"8ff9f13616d90f032f92bb0f80c18039a599de6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2023-04-01 02:24:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4266daaf5dd5e3a28ac784d30324eb7d58cbe577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 02:44:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1889ea4c66e447db7d1594d3c61bc32\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/773c7ade1cfb4126bd0e1f0c5dd9e2e4 : SUCCESS in 2m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26a641c2e61b4211b31241deb8c2ba99 : SUCCESS in 7m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8bce6158222e43bcb22b7e2dfd04e6f2 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d525abd46900b5146348b93cf69a5d11774d8f09","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-01 02:48:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec010995c4d7bb163ccfe879e8eaf11df35cb7f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 02:48:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3402d954785e25bd4c3733c0183b3b8ce35839b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 02:56:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c532a7feec0d470badf8a2c610062d9f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3aea3e7fc1fc4ac792865bb65d18aa8a : SUCCESS in 7m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0383edc270864e97b1ff4b60aa4bb755 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"585efac03ceddf8bb7c9479f5688a5b1b38d3b0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 02:56:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47a43e5ba86247b39989e87bda33f1b9d5a101f5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 03:22:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2257ca66b94ad2a889431f61ec6d33\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/26ed1e243f6a4a90b62ea0d6724e5623 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e4748ca35008a4c25e514ff6a3e941edb00c809","revisions":{"6e4748ca35008a4c25e514ff6a3e941edb00c809":{"kind":"REWORK","_number":1,"created":"2023-04-01 02:24:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/879213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/13/879213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/13/879213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/13/879213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/13/879213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/13/879213/1"}}},"commit":{"parents":[{"commit":"a402e1ca97df4198ffef3c408f304ce4f8e2b97a","subject":"Merge \"Add virtualization (virt) repo to StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a402e1ca97df4198ffef3c408f304ce4f8e2b97a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-01 02:24:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2023-04-01 02:24:20.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I8ba595466ebc9e0546680be0c16454d9e7903d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
