)]}'
{"id":"openstack%2Fproject-config~879319","triplet_id":"openstack%2Fproject-config~master~I9d99ba7edb996e871e8bc8d2cc1c3db0197a4b9e","project":"openstack/project-config","branch":"master","topic":"fix-ptl-approved","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-04-03 18:20:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d99ba7edb996e871e8bc8d2cc1c3db0197a4b9e","subject":"Fix PTL-Approved column in RelMgt dashboard","status":"MERGED","created":"2023-04-03 09:48:30.000000000","updated":"2023-04-03 18:41:56.000000000","submitted":"2023-04-03 18:20:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879319-fix-ptl-approved","meta_rev_id":"0d7380caf05f769b2e51be6e53a27195cd0b84f9","_number":879319,"virtual_id_number":879319,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 18:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-04-03 17:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-03 17:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-04-03 18:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 10:00:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 12:42:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-04-03 12:57:03.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-04-03 17:13:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"c9c786d52e29593c2036576c667128f05a17605d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-03 09:48:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f409fba7ef8078af21c973c1dd327f4d5a24cc2d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-03 09:49:00.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bec930685d7d541b9d295dcc407a086c1469a0f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:00:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46ef3730fc364a1588ab89ffc72109d0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ae4449f7c6a345dcabe4f849f9dbf844 : SUCCESS in 2m 36s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ef848b80a994035a514682944df3ab8 : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/483688ba64f14950aa609247142f607c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5b476e1b2764e9db308a1247a12569d78809d216","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-03 12:42:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b65231984b3dc057a62004ddb0de245160a4f6cc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-03 12:49:33.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"a3516d81ab8a184eb4916ced9e1d8c1beaf61488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-03 12:55:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a04a70bf4573b9a504b2f70e7602a1ce80d9da2b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-03 12:57:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aaf0b67f2cc1f7a586dd4b687bf8500f51640bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 13:04:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6475fdc09ab047aba4ebec61769de254\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/87befb225538448291912f23a4cb5c16 : SUCCESS in 3m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a28f3ed37dd14ee8a80e4d59e47ff528 : SUCCESS in 6m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/73d192d8d94e47f9b4413a51f67e5d6b : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d496e16f1c4a3c66d9b931bdfdc11355b120747b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-03 16:47:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a121b8774e0aed54ae2885a3a22313dfc9076100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-04-03 16:48:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a9ae6a50f955ca190442bdf566e21513b25790d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:02:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d4d24c60014369baebbbd646d44558\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ec40f29a2b63423e90b010c9a151c89d : SUCCESS in 3m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42d4fbffaa8e4faba7d2050637c669ca : SUCCESS in 7m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/97338372f5094a65bddf40aff3bf6e7d : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"b827db4a53840c96ac81a0a065f9384539151d35","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-03 17:08:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ffb6156c5afcb4fc77b9090c5fe8b74f8835c19a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-03 17:13:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c1aedf3915fd53509e83ff59dabc1636fffe0b10","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-03 18:07:42.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4278c36bd383bb09429a721779028e86a5835ccb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 18:08:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7fe859e1bc5dd9583290adc309eefbb5de5e0852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 18:20:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1302af890474643a1eb876b40e82497\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/450aa25a1e8d47619651ac78d049ddc3 : SUCCESS in 6m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0c3470300661425bb0f5b428e8553bdf : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":4},{"id":"faaba5cb10bd0efebbe48e69210612dacaa9bc0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 18:20:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0d7380caf05f769b2e51be6e53a27195cd0b84f9","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 18:41:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec69d42776f44598c4f9d2eb1c292bb\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/80ca6d86d32342a8878e1faf2688b449 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15d80f34ae4933d0751d9c35e6d36eb2cfc7965e","revisions":{"eec88d2bfb0e2445fb3b042da22e3478bfda2a18":{"kind":"REWORK","_number":1,"created":"2023-04-03 09:48:30.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/19/879319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/879319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/879319/1"}}},"commit":{"parents":[{"commit":"6e4748ca35008a4c25e514ff6a3e941edb00c809","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 09:40:04.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 09:46:58.000000000","tz":120},"subject":"Fix PTL-Approved column in RelMgt dashboard","message":"Fix PTL-Approved column in RelMgt dashboard\n\nThe PTL-Approved label is currently set as a submit requirement, so that\nit appears in the review dashboards. However, in order to not block the\nsubmit, that requirement is always satisfied (the default of 0 is above\nMIN), so that column is always shown with a green \"Satisfied\" checkmark.\n\nThis change makes PTL-Approved a blocking submit requirement (now\nneeding the MAX of +1) while allowing the release managers to set +1\nin extreme cases where PTL approval cannot be obtained. This should\nensure that the green checkmark in the PTL-Approved column works\ncorrectly and lets us spot which changes can be processed quickly.\n\nChange-Id: I9d99ba7edb996e871e8bc8d2cc1c3db0197a4b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eec88d2bfb0e2445fb3b042da22e3478bfda2a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eec88d2bfb0e2445fb3b042da22e3478bfda2a18"}]},"branch":"refs/heads/master"},"430c64890bea6cb785fef3ee741769bb6c3e4252":{"kind":"REWORK","_number":2,"created":"2023-04-03 12:55:16.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/19/879319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/879319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/879319/2"}}},"commit":{"parents":[{"commit":"6e4748ca35008a4c25e514ff6a3e941edb00c809","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 09:40:04.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 12:53:15.000000000","tz":120},"subject":"Fix PTL-Approved column in RelMgt dashboard","message":"Fix PTL-Approved column in RelMgt dashboard\n\nThe PTL-Approved label is currently set as a submit requirement, so that\nit appears in the review dashboards. However, in order to not block the\nsubmit, that requirement is always satisfied (the default of 0 is above\nMIN), so that column is always shown with a green \"Satisfied\" checkmark.\n\nThis change makes PTL-Approved a blocking submit requirement (the\ndefault of 0 now being the MIN) while allowing the release managers to\nset +1 in extreme cases where PTL approval cannot be obtained. This\nshould ensure that the green checkmark in the PTL-Approved column works\ncorrectly and lets us spot which changes can be processed quickly.\n\nChange-Id: I9d99ba7edb996e871e8bc8d2cc1c3db0197a4b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/430c64890bea6cb785fef3ee741769bb6c3e4252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/430c64890bea6cb785fef3ee741769bb6c3e4252"}]},"branch":"refs/heads/master"},"c224d580c310ea403feca4d626432f00258cfc57":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-03 16:47:45.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/19/879319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/879319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/879319/3"}}},"commit":{"parents":[{"commit":"24626158149c04e4ef594140482bd90f474b0a6d","subject":"Require function \u003d NoBlock to be set on Gerrit labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24626158149c04e4ef594140482bd90f474b0a6d"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 09:40:04.000000000","tz":120},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 16:47:38.000000000","tz":-420},"subject":"Fix PTL-Approved column in RelMgt dashboard","message":"Fix PTL-Approved column in RelMgt dashboard\n\nThe PTL-Approved label is currently set as a submit requirement, so that\nit appears in the review dashboards. However, in order to not block the\nsubmit, that requirement is always satisfied (the default of 0 is above\nMIN), so that column is always shown with a green \"Satisfied\" checkmark.\n\nThis change makes PTL-Approved a blocking submit requirement (the\ndefault of 0 now being the MIN) while allowing the release managers to\nset +1 in extreme cases where PTL approval cannot be obtained. This\nshould ensure that the green checkmark in the PTL-Approved column works\ncorrectly and lets us spot which changes can be processed quickly.\n\nChange-Id: I9d99ba7edb996e871e8bc8d2cc1c3db0197a4b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c224d580c310ea403feca4d626432f00258cfc57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c224d580c310ea403feca4d626432f00258cfc57"}]},"branch":"refs/heads/master"},"15d80f34ae4933d0751d9c35e6d36eb2cfc7965e":{"kind":"REWORK","_number":4,"created":"2023-04-03 16:48:21.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/19/879319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/879319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/879319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/879319/4"}}},"commit":{"parents":[{"commit":"6e4748ca35008a4c25e514ff6a3e941edb00c809","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 09:40:04.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-04-03 16:48:04.000000000","tz":120},"subject":"Fix PTL-Approved column in RelMgt dashboard","message":"Fix PTL-Approved column in RelMgt dashboard\n\nThe PTL-Approved label is currently set as a submit requirement, so that\nit appears in the review dashboards. However, in order to not block the\nsubmit, that requirement is always satisfied (the default of 0 is above\nMIN), so that column is always shown with a green \"Satisfied\" checkmark.\n\nThis change makes PTL-Approved a blocking submit requirement (the\ndefault of 0 now being the MIN) while allowing the release managers to\nset +1 in extreme cases where PTL approval cannot be obtained. This\nshould ensure that the green checkmark in the PTL-Approved column works\ncorrectly and lets us spot which changes can be processed quickly.\n\nChange-Id: I9d99ba7edb996e871e8bc8d2cc1c3db0197a4b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15d80f34ae4933d0751d9c35e6d36eb2cfc7965e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/15d80f34ae4933d0751d9c35e6d36eb2cfc7965e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
