)]}'
{"id":"openstack%2Fproject-config~879364","triplet_id":"openstack%2Fproject-config~master~I9e83c1cde3fe20ea2c34cdf86cd2fd3006bfe62a","project":"openstack/project-config","branch":"master","topic":"require-function-noblock","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-04-03 21:29:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e83c1cde3fe20ea2c34cdf86cd2fd3006bfe62a","subject":"Require function \u003d NoBlock to be set on Gerrit labels","status":"MERGED","created":"2023-04-03 16:25:52.000000000","updated":"2023-04-03 21:29:20.000000000","submitted":"2023-04-03 21:29:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879364-require-function-noblock","meta_rev_id":"19a5adaa1d814563215e911ee0fba9f0de9008bc","_number":879364,"virtual_id_number":879364,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 21:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-03 21:17:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2023-04-03 17:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-03 21:17:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 16:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 17:11:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-04-03 21:17:13.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"4634f2fcddfd61e880caafcae584daad450bea1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-03 16:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45b6490d041b350a93ec8220fb30bad7886c8916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:39:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364c2d20b2d54bd5ab3aec45b7c5fdb0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/25eaa5b2db6743289146afb1d0161c6e : SUCCESS in 4m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8fc422a4e1c34fb889302e0917f2fa42 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4a2455f9d0cbb6d8b924732396184e8369dd56fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-03 16:53:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17857478e9f2ba12f3d16f5c976ddb3718119e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:02:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb1193d1c17344bdbb4d3ef2972c24f8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9d9804cefb3942bf93cd53afb8f0e738 : SUCCESS in 3m 10s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/34957e6a0e4e49ca9f98f735c322ea19 : SUCCESS in 6m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/56fb606e751e46e89d2ef4e42bf1d1c6 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2961e0ceb97f757f98503f727d1c91be85b3a46c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-03 17:07:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db4ed20cac7e3740f5746f05cf46409ce1f5cb82","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-03 17:11:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c05216212106c57d71968b2c45ed8cb70c1507d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:15:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c1425e9f57f46c0866b104bfd00bfc3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d08297168d8e42e4b6da81de1d3b466b : SUCCESS in 2m 56s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a38433a14e3548708b26720e2c27f1c4 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7293a96fac1a640c724152cbffa14dc3b3ecbb42","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-03 21:17:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"361ec99369d240e2310d04446abb1c1fb9d3e5a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 21:17:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f11cf7d9029726c23d3a64f1fb3cf2141dccd496","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 21:29:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/379c1890b98646c6ac14c74cdeb69415\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a151d2bab1864f60aae7a630048332a2 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"19a5adaa1d814563215e911ee0fba9f0de9008bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 21:29:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5fc22f27cfcdb7b99dc32880569c290a8e9391a8","revisions":{"24626158149c04e4ef594140482bd90f474b0a6d":{"kind":"REWORK","_number":1,"created":"2023-04-03 16:25:52.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/64/879364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/879364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/879364/1"}}},"commit":{"parents":[{"commit":"6e4748ca35008a4c25e514ff6a3e941edb00c809","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 16:23:17.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 16:23:17.000000000","tz":-420},"subject":"Require function \u003d NoBlock to be set on Gerrit labels","message":"Require function \u003d NoBlock to be set on Gerrit labels\n\nA recent change to the openstack/releases ACLs pointed out that we don\u0027t\nrequire function to be set on Gerrit label definitions. This would\nresult in the Gerrit default of MaxWithBlock which will interfere with\nsubmit requirements.\n\nEnforce that function is set and that the value is NoBlock via our\nnormalize script. This will add function \u003d NoBlock entries to the file\nif not set which results in a diff causing the test to fail.\n\nIn order to do this I refactored the submit-requirements and function\nchecking of the normalize script a bit. We now check the label section\nindependently of all other sections which allows us to reduce repetition\nwhen dealing with label sections.\n\nChange-Id: I9e83c1cde3fe20ea2c34cdf86cd2fd3006bfe62a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24626158149c04e4ef594140482bd90f474b0a6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24626158149c04e4ef594140482bd90f474b0a6d"}]},"branch":"refs/heads/master"},"9d3015069b7d18e93891746d1c32d12d470f1a8c":{"kind":"REWORK","_number":2,"created":"2023-04-03 16:53:55.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/64/879364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/879364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/879364/2"}}},"commit":{"parents":[{"commit":"6e4748ca35008a4c25e514ff6a3e941edb00c809","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 16:23:17.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 16:53:49.000000000","tz":-420},"subject":"Require function \u003d NoBlock to be set on Gerrit labels","message":"Require function \u003d NoBlock to be set on Gerrit labels\n\nA recent change to the openstack/releases ACLs pointed out that we don\u0027t\nrequire function to be set on Gerrit label definitions. This would\nresult in the Gerrit default of MaxWithBlock which will interfere with\nsubmit requirements.\n\nEnforce that function is set and that the value is NoBlock via our\nnormalize script. This will add function \u003d NoBlock entries to the file\nif not set which results in a diff causing the test to fail.\n\nIn order to do this I refactored the submit-requirements and function\nchecking of the normalize script a bit. We now check the label section\nindependently of all other sections which allows us to reduce repetition\nwhen dealing with label sections.\n\nChange-Id: I9e83c1cde3fe20ea2c34cdf86cd2fd3006bfe62a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d3015069b7d18e93891746d1c32d12d470f1a8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d3015069b7d18e93891746d1c32d12d470f1a8c"}]},"branch":"refs/heads/master"},"5fc22f27cfcdb7b99dc32880569c290a8e9391a8":{"kind":"REWORK","_number":3,"created":"2023-04-03 17:07:17.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/64/879364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/64/879364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/64/879364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/64/879364/3"}}},"commit":{"parents":[{"commit":"6e4748ca35008a4c25e514ff6a3e941edb00c809","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e4748ca35008a4c25e514ff6a3e941edb00c809"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 16:23:17.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-04-03 17:07:07.000000000","tz":-420},"subject":"Require function \u003d NoBlock to be set on Gerrit labels","message":"Require function \u003d NoBlock to be set on Gerrit labels\n\nA recent change to the openstack/releases ACLs pointed out that we don\u0027t\nrequire function to be set on Gerrit label definitions. This would\nresult in the Gerrit default of MaxWithBlock which will interfere with\nsubmit requirements.\n\nEnforce that function is set and that the value is NoBlock via our\nnormalize script. This will add function \u003d NoBlock entries to the file\nif not set which results in a diff causing the test to fail.\n\nIn order to do this I refactored the submit-requirements and function\nchecking of the normalize script a bit. We now check the label section\nindependently of all other sections which allows us to reduce repetition\nwhen dealing with label sections.\n\nChange-Id: I9e83c1cde3fe20ea2c34cdf86cd2fd3006bfe62a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5fc22f27cfcdb7b99dc32880569c290a8e9391a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5fc22f27cfcdb7b99dc32880569c290a8e9391a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
