)]}'
{"id":"openstack%2Fproject-config~879456","triplet_id":"openstack%2Fproject-config~master~I72f9d1213c52a83cb3a9c29ce29f470023e96c0e","project":"openstack/project-config","branch":"master","topic":"gerrit-3.7","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-04-06 23:58:13.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-04-06 23:58:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72f9d1213c52a83cb3a9c29ce29f470023e96c0e","subject":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent","status":"MERGED","created":"2023-04-04 13:40:52.000000000","updated":"2023-04-07 00:14:20.000000000","submitted":"2023-04-06 23:58:13.000000000","submitter":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879456-gerrit-3.7","meta_rev_id":"6a4618183ec8436089f5b72c1f9c5d22d875c8ef","_number":879456,"virtual_id_number":879456,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-04-06 23:58:13.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-04 22:22:47.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-04-04 22:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-05 20:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-04-06 23:58:13.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":1,"date":"2023-04-05 07:57:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-04 22:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":2,"date":"2023-04-06 17:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2023-04-06 23:58:13.000000000","_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}],"CC":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-04 13:47:56.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-04-04 13:48:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-04-04 13:49:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 14:08:35.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2023-04-04 14:47:14.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-04-04 15:06:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-04 22:22:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-04 22:23:33.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"CC"},{"updated":"2023-04-05 07:57:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-05 20:44:25.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-04-06 23:58:13.000000000","updated_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"reviewer":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"state":"REVIEWER"}],"messages":[{"id":"7d82fafb76674e4cf9d532e6657ad86a51ee45a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-04-04 13:40:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9866076ce27f075e2ef02a7b231f377fb10713b6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-04 13:47:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62879b819d23687f1a935788572e4bac45eecf3a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-04 13:48:25.000000000","message":"Patch Set 1:\n\nMake sure you add this to https://wiki.openstack.org/wiki/Meetings/InfraTeamMeeting#Upcoming_Project_Renames if you want it included in the upcoming project rename maintenance. Also I recommend dropping the Depends-On and making the governance change depend on this one instead, so that it doesn\u0027t block us from merging during the maintenance window.","accounts_in_message":[],"_revision_number":1},{"id":"d7e669f2ba509eb6609143a81d8e1e4b94940577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 13:49:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa08bfc0645d434797b4d56a091fb67d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ce2ab20c9584430386b297faca66f8fe : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/39aa5db69c1146b8985f805ee09ebe71 : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/66b3cc7c7658403c930130fb6a7219f3 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"18c28652baafa589ca5480ee766edd64f806364f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-04-04 13:55:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e89e3b9517d281f42d5ab14ad8f821ca1559ac5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 14:03:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ba0514b6f74f02a20d5858a764a329\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a3987ab3eda5456093aaa3c04cdf70e4 : SUCCESS in 3m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bf0a65740960408dbf940ecc0a5753f8 : SUCCESS in 5m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ecf2fb7544e8447f85d88c87dac0686f : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"dd812b8f3e388d54863558f23dfcd200543217ca","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-04-04 14:08:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ee7b58de00c8e2db3ec6e877887024db9fcb6d0a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-04 14:47:14.000000000","message":"Patch Set 2: Code-Review+2\n\nDon\u0027t forget to add it to the meeting agenda linked from my earlier comment.","accounts_in_message":[],"_revision_number":2},{"id":"e582edbfec96c88fba4437adc305be315930a392","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-04 15:06:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4e5858b4667271f180c133238a6e8e2688fbf40f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-04-04 15:38:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d399153a6b77cfa1d4ed1cfab5cb50cf726712aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-04 22:22:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"925c4e9f74d399c8e7ebec40392751f511afe5da","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-04 22:22:42.000000000","message":"Patch Set 3:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"0b4860f89036f3a0c73c06cc697e3f4b607ff0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 22:22:47.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"f7d85be40121e4420268fd607782e8bd047327b7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-04 22:23:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d25cf75a91bff2f582dc23fd1cdc5ff4ab26199b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-05 07:57:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"497bf8ded8112464d5c88e3a384737257279622c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-05 20:44:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d58bf01c5b4ceefe8662e2b6109eb493f951ef5f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-06 17:48:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8cabd296dab4dc9756bf955305b35e01a00e4944","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2023-04-06 23:58:13.000000000","message":"Patch Set 3: Verified+2 Workflow+1 Code-Review+2\n\nBypassing Zuul for rename","accounts_in_message":[],"_revision_number":3},{"id":"db13c694f89db6a8b629868733ba169063855965","tag":"autogenerated:gerrit:merged","author":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"},"date":"2023-04-06 23:58:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6a4618183ec8436089f5b72c1f9c5d22d875c8ef","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 00:14:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dce329ba4314d3e8924dfa7ab199196\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/9891dc34ec3a4406b7eb34d39c7c91ff : SUCCESS in 7m 53s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/2766e544f2794b5289758514c7ff1f28 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"10aba3c679cbe1f5a9ae5dd66c569f95a9334fb5","revisions":{"03ba83c466e068e144d94a0a6739f99dbbea31af":{"kind":"REWORK","_number":1,"created":"2023-04-04 13:40:52.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/56/879456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/879456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/879456/1"}}},"commit":{"parents":[{"commit":"d35367ededf6671ab982da4277ecdee771c2e0de","subject":"Merge \"Require function \u003d NoBlock to be set on Gerrit labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d35367ededf6671ab982da4277ecdee771c2e0de"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-04-04 13:36:45.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-04-04 13:40:38.000000000","tz":60},"subject":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent","message":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent\n\nDuring the Bobcat PTG it was discussed and accepted to move the\novn-bgp-agent project under the Neutron governance [0].\n\nThis patch is just renaming the x/ references to the openstack/ namespace\nfor the ovn-bgp-agent project.\n\n[0] https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033161.html\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/879455\nChange-Id: I72f9d1213c52a83cb3a9c29ce29f470023e96c0e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03ba83c466e068e144d94a0a6739f99dbbea31af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03ba83c466e068e144d94a0a6739f99dbbea31af"}]},"branch":"refs/heads/master"},"05a96d793056e777ee88cd4229110448af3811db":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-04 13:55:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/56/879456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/879456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/879456/2"}}},"commit":{"parents":[{"commit":"d35367ededf6671ab982da4277ecdee771c2e0de","subject":"Merge \"Require function \u003d NoBlock to be set on Gerrit labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d35367ededf6671ab982da4277ecdee771c2e0de"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-04-04 13:36:45.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-04-04 13:54:41.000000000","tz":60},"subject":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent","message":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent\n\nDuring the Bobcat PTG it was discussed and accepted to move the\novn-bgp-agent project under the Neutron governance [0].\n\nThis patch is just renaming the x/ references to the openstack/ namespace\nfor the ovn-bgp-agent project.\n\n[0] https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033161.html\n\nChange-Id: I72f9d1213c52a83cb3a9c29ce29f470023e96c0e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05a96d793056e777ee88cd4229110448af3811db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/05a96d793056e777ee88cd4229110448af3811db"}]},"branch":"refs/heads/master"},"10aba3c679cbe1f5a9ae5dd66c569f95a9334fb5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-04 22:22:05.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/56/879456/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/879456/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/879456/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/879456/3"}}},"commit":{"parents":[{"commit":"62f9137743c1856c6484eee3a718637c88090f0c","subject":"Rename x/xstatic-angular-fileupload-\u003eopenstack/xstatic-angular-fileupload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/62f9137743c1856c6484eee3a718637c88090f0c"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-04-04 13:36:45.000000000","tz":60},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-04 22:21:39.000000000","tz":600},"subject":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent","message":"Rename x/ovn-bgp-agent to openstack/ovn-bgp-agent\n\nDuring the Bobcat PTG it was discussed and accepted to move the\novn-bgp-agent project under the Neutron governance [0].\n\nThis patch is just renaming the x/ references to the openstack/ namespace\nfor the ovn-bgp-agent project.\n\n[0] https://lists.openstack.org/pipermail/openstack-discuss/2023-April/033161.html\n\nChange-Id: I72f9d1213c52a83cb3a9c29ce29f470023e96c0e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10aba3c679cbe1f5a9ae5dd66c569f95a9334fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10aba3c679cbe1f5a9ae5dd66c569f95a9334fb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32627,"name":"ianw.admin","username":"ianw.admin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
