)]}'
{"id":"openstack%2Fproject-config~879469","triplet_id":"openstack%2Fproject-config~master~I71010276b9b35cb67e3717002377707e1435f9f9","project":"openstack/project-config","branch":"master","topic":"cinder-nfs","attention_set":{},"removed_from_attention_set":{"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2023-05-19 13:14:05.000000000","reason":"Change was submitted"},"34059":{"account":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"last_update":"2023-05-19 13:14:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71010276b9b35cb67e3717002377707e1435f9f9","subject":"Add cinder-nfs charm to OpenStack charms","status":"MERGED","created":"2023-04-04 15:33:54.000000000","updated":"2023-05-19 13:30:11.000000000","submitted":"2023-05-19 13:14:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"879469-cinder-nfs","meta_rev_id":"10a79c41ea1c96502d905a6249dbb96eedc23302","_number":879469,"virtual_id_number":879469,"owner":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 13:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2023-05-17 11:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-05-19 02:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":1,"date":"2023-05-16 18:24:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2023-04-26 21:35:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-02 11:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-19 13:05:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-04 15:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-05 08:41:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2023-04-26 21:35:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-05-02 11:58:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-05-16 18:24:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-05-17 11:13:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-17 15:34:02.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"}],"messages":[{"id":"5361e9b24aeff274e6555b4d069e6724cece2286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"date":"2023-04-04 15:33:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3757251ebd1631acdbd353f97a85ea21524a125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 15:41:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173a08db83e044389dc7e4e771072b1e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6d43b5e2da3b41b29471ffefd72d6a87 : SUCCESS in 2m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f857fad2a4ef4785ae44e7aa5d06a4fe : SUCCESS in 5m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/78977c054eb541d484adcf5815f987fa : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"416fc4fd3cb3484e03053af645e15a5c47713557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"date":"2023-04-04 17:10:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81a7d02d9b677e8900e2aeb145556f1e41ca3994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:19:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bf169ad458b47da8022779fce633539\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bfd5ec3c37a24f90ba5022da6965accc : SUCCESS in 2m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/91251951232c4da6a9553e039149a8af : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a0303871716e4426aaccbc090f0eb20b : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cc3d03f6ebf054227c25a974a8a8bddec8a1cb23","author":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"date":"2023-04-04 18:12:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa3ea62492ba46c71d8d6ba023931839088120ea","author":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"date":"2023-04-04 22:08:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efc9f5365be9ff43228641f847b1dd3519f9f7c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"date":"2023-04-05 00:14:53.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"626c234188e51bb83a6793f6e1125f9e63b56290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 00:23:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c0de187c494cbba94a9af642f8a30f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4da3d910d26941b58ec01e0950073144 : SUCCESS in 2m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e251da9908ce41b88f998120288bd068 : SUCCESS in 4m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9c91cafdc5bc4fc1a8549d2ff0b8180f : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"848946dead8787124a338cc8fa97a801a18425f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-04-05 08:41:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"facb113af598ca252275cc856f8c2cf7131cdd0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"date":"2023-04-05 13:45:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a1f0f4b3dcd0f44068e81cf13e5a081bf1441b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 14:02:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ee04cee2c243c9ab44f24df40ce734\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6728ec120f14457aaaf18007221adb50 : SUCCESS in 3m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5a2dc5b932ce42048c59d6b8692b4597 : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eb0711b6428e43849961478e71f6f718 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"e15ba15f006b264787a3e37a349dfb5d0f5a87f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-26 21:35:21.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9034c00722f260fec1a69824ac22fb754b85c518","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 11:58:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"35ba1c240b7da26a2437103a8562108797649122","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-16 18:24:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e82ff7c37dbc8a182d763ababa374470c56163e0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-16 19:01:13.000000000","message":"Patch Set 4:\n\nStill need a +1 from Felipe to confirm or a link to a similarly acknowledged openstack/governance change for this.","accounts_in_message":[],"_revision_number":4},{"id":"966189807e8ea13b3896ff76534614c3207cb283","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 11:13:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e9bebb63e056c87e59e427bb9370544b5c9d93d6","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-05-19 02:36:58.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be475a5f21eb70b10ef24432a45ad7373808369e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-19 13:05:41.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"55e57318028a4afe51c390324807841a25549b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:06:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3f5783d53205fee083eb3ca26ade0d5d2223754a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:14:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2d9860564f74f7587c9fb6ab4cb3919\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4e8e6051b89648598a9e9886d870537e : SUCCESS in 6m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ab4b700841834a75aa8d2819c95ec5b0 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"20ebc10d44905072b8a89dcd49016a84212cd838","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:14:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"10a79c41ea1c96502d905a6249dbb96eedc23302","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:30:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2675234546424e23ae4f53efea9b1979\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/4370066e0f124ed693efec53b99aa0eb : SUCCESS in 5m 17s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/7fbb2ca9e1c84b13b15ba602304b837f : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d88083309f951182d20a2df9725559631dc71aef","revisions":{"e1fb68e634e03a1cf5ac1cfb2d10b62821e78f2d":{"kind":"REWORK","_number":1,"created":"2023-04-04 15:33:54.000000000","uploader":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"ref":"refs/changes/69/879469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/879469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/879469/1"}}},"commit":{"parents":[{"commit":"d35367ededf6671ab982da4277ecdee771c2e0de","subject":"Merge \"Require function \u003d NoBlock to be set on Gerrit labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d35367ededf6671ab982da4277ecdee771c2e0de"}]}],"author":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-04 15:24:30.000000000","tz":0},"committer":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-04 15:30:01.000000000","tz":0},"subject":"Add cinder-nfs charm to OpenStack charms","message":"Add cinder-nfs charm to OpenStack charms\n\nAdd new charm that provides the NFS backend support to the OpenStack\nCinder service.\n\nChange-Id: I71010276b9b35cb67e3717002377707e1435f9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1fb68e634e03a1cf5ac1cfb2d10b62821e78f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1fb68e634e03a1cf5ac1cfb2d10b62821e78f2d"}]},"branch":"refs/heads/master"},"df0e658410c15ac68ff959ffd40b2bbf73e0c203":{"kind":"REWORK","_number":2,"created":"2023-04-04 17:10:59.000000000","uploader":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"ref":"refs/changes/69/879469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/879469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/879469/2"}}},"commit":{"parents":[{"commit":"d35367ededf6671ab982da4277ecdee771c2e0de","subject":"Merge \"Require function \u003d NoBlock to be set on Gerrit labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d35367ededf6671ab982da4277ecdee771c2e0de"}]}],"author":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-04 15:24:30.000000000","tz":0},"committer":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-04 17:10:43.000000000","tz":0},"subject":"Add cinder-nfs charm to OpenStack charms","message":"Add cinder-nfs charm to OpenStack charms\n\nAdd new charm that provides the NFS backend support to the OpenStack\nCinder service.\n\nChange-Id: I71010276b9b35cb67e3717002377707e1435f9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df0e658410c15ac68ff959ffd40b2bbf73e0c203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df0e658410c15ac68ff959ffd40b2bbf73e0c203"}]},"branch":"refs/heads/master"},"b6f7973dd280d4447d2e16dc9048d72969c4dbfc":{"kind":"NO_CHANGE","_number":3,"created":"2023-04-05 00:14:53.000000000","uploader":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"ref":"refs/changes/69/879469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/879469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/879469/3"}}},"commit":{"parents":[{"commit":"d35367ededf6671ab982da4277ecdee771c2e0de","subject":"Merge \"Require function \u003d NoBlock to be set on Gerrit labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d35367ededf6671ab982da4277ecdee771c2e0de"}]}],"author":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-04 15:24:30.000000000","tz":0},"committer":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-05 00:14:34.000000000","tz":0},"subject":"Add cinder-nfs charm to OpenStack charms","message":"Add cinder-nfs charm to OpenStack charms\n\nAdd new charm that provides the NFS backend support to the OpenStack\nCinder service.\n\nChange-Id: I71010276b9b35cb67e3717002377707e1435f9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6f7973dd280d4447d2e16dc9048d72969c4dbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6f7973dd280d4447d2e16dc9048d72969c4dbfc"}]},"branch":"refs/heads/master"},"d88083309f951182d20a2df9725559631dc71aef":{"kind":"REWORK","_number":4,"created":"2023-04-05 13:45:49.000000000","uploader":{"_account_id":34059,"name":"Andy Wu","display_name":"Andy Wu","email":"andy.wu@canonical.com","username":"andywu"},"ref":"refs/changes/69/879469/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/69/879469/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/69/879469/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/69/879469/4"}}},"commit":{"parents":[{"commit":"d35367ededf6671ab982da4277ecdee771c2e0de","subject":"Merge \"Require function \u003d NoBlock to be set on Gerrit labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d35367ededf6671ab982da4277ecdee771c2e0de"}]}],"author":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-04 15:24:30.000000000","tz":0},"committer":{"name":"Andy Wu","email":"andy.wu@canonical.com","date":"2023-04-05 13:45:36.000000000","tz":0},"subject":"Add cinder-nfs charm to OpenStack charms","message":"Add cinder-nfs charm to OpenStack charms\n\nAdd new charm that provides the NFS backend support to the OpenStack\nCinder service.\n\nChange-Id: I71010276b9b35cb67e3717002377707e1435f9f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d88083309f951182d20a2df9725559631dc71aef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d88083309f951182d20a2df9725559631dc71aef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
