)]}'
{"id":"openstack%2Fproject-config~880115","triplet_id":"openstack%2Fproject-config~master~I1f11c07e3786bd1a68b43d908d939fde42ddb99c","project":"openstack/project-config","branch":"master","topic":"nocodechange","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-04-19 23:08:50.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-04-20 01:52:51.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I1f11c07e3786bd1a68b43d908d939fde42ddb99c","subject":"gerrit/acls : add NO_CODE_CHANGE","status":"MERGED","created":"2023-04-12 01:39:06.000000000","updated":"2023-04-20 02:34:18.000000000","submitted":"2023-04-20 02:00:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880115-nocodechange","meta_rev_id":"3cc30e1c229dac99a957aa8dd030bc71a8d2ac0d","_number":880115,"virtual_id_number":880115,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 02:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-04-20 00:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-04-20 00:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 01:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 01:56:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 07:15:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2023-04-13 17:33:32.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2023-04-19 23:08:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"d98bc6733b3a62556bc98daa90551b2a9890b554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-12 01:39:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b94f82101ea0e2037733c22efcafd1b9dae64d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 01:56:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ced225ff5849098d8639fc8b4a8ea4\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/954c7c5456b3432da824d743dad393b3 : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9050197a02b24ff9ba570fef46752e8e : SUCCESS in 5m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3228d182750445ccb84c0c1e53342825 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6055068210d652b02aa80f4491add8356f59e744","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-04-13 07:15:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"25c55740fcded1a10cdef5e9b0b2dff00f38b79b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-13 17:33:32.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"448ab4cf131f9fd9add13dcfe6d05ec3780a2314","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-14 02:06:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c33a08ba194f94c8a65ff94db0158496e056b6f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-19 05:05:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4ab3de87c3d1687c39817f799357fd738de16cf4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-19 05:06:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9698e215db4f1a4324a01001e134e9dc40cc3af4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 05:16:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b721883829b843429d7fdb4c3b62f091\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b4e153cd9252410a952d9f110d7411cc : SUCCESS in 3m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e1bb5271f60b415cb7b88a68cd348f00 : SUCCESS in 6m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1532bca5394c4318a4a7a63a7f10ff48 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"42290affd2561f307b281e456ca15a0b6931a9f1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-19 05:26:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e839504477bfa8cce9576404e16a699c049822ea","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-19 23:08:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"787a05d0030e2431992db1ea803076005f45b6f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-20 00:56:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8268b4e398cd1944d0f7a8a29b4f01810cb7466e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 01:03:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/873add4b2bb94321af263dd160198fa1\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c48b7d4dee8744c5ab6a7759a129e781 : SUCCESS in 3m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5a79b5702d1f47f29602e1be9eb88db8 : SUCCESS in 5m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/63c665df4424491b9609a6450a3833ea : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d5d1b8ee014b25a8df1e7f13eca06ae6ddfaa61d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-20 01:52:51.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"690333fd2fc97ffb0ea6d1edfacecc77cbf5e2a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 01:53:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"746ea73553837ff7c7eeb34bc65a7073eb903d47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 02:00:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d431e26216d641a39d0b1a64431c3475\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1bf5d176fa424805b2c575d193b4659f : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cf0e861a0cef4dc389f4be000df095c9 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"463e29454578ad5a9b117333496e9c14104b4937","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 02:00:29.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"3cc30e1c229dac99a957aa8dd030bc71a8d2ac0d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 02:34:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b84010fa0d44758529927079fc5478\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0a5049733727450cb1201f755bd4be4d : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6f70d8914590a290adf0c3947a96940fb6cb1b66","revisions":{"7f90fec609c4d7e4b1145d81e806c00f0c66f61e":{"kind":"REWORK","_number":1,"created":"2023-04-12 01:39:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/15/880115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/880115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/880115/1"}}},"commit":{"parents":[{"commit":"c1e7756e025e96b8ad41067e28985ae214954bd8","subject":"Merge \"Stop using Storyboard for Magnum projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1e7756e025e96b8ad41067e28985ae214954bd8"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-12 01:11:39.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-12 01:23:26.000000000","tz":600},"subject":"gerrit/acls : add NO_CODE_CHANGE","message":"gerrit/acls : add NO_CODE_CHANGE\n\nThe old copyAllScoresIfNoChange flag defaults to true, so any ACLs\nwithout it explicitly set to \"false\" (i.e. all of ours) have had the\nequivalant copy-condition flag \"changekind:NO_CODE_CHANGE\" added\nduring migration (see [1]).\n\nAny ACL\u0027s that were not specifying changekind:NO_CODE_CHANGE are\nupdated so that we are in sync with what is committed in Gerrit.\n\nWe know that updates that match changekind:TRIVIAL_REBASE also match\nchangekind:NO_CODE_CHANGE; however the migration has still added the\nlatter to all rules.  Thus to avoid confusion, we also add it here.\n\n[1] https://bugs.chromium.org/p/gerrit/issues/detail?id\u003d16839\n\nChange-Id: I1f11c07e3786bd1a68b43d908d939fde42ddb99c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f90fec609c4d7e4b1145d81e806c00f0c66f61e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f90fec609c4d7e4b1145d81e806c00f0c66f61e"}]},"branch":"refs/heads/master"},"66c99e87df21802bd1239312785dbe53f5794e9f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-19 05:05:33.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/15/880115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/880115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/880115/2"}}},"commit":{"parents":[{"commit":"76e73ab492635bf64e957d8a498f36396d148128","subject":"Indent Gerrit ACL options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76e73ab492635bf64e957d8a498f36396d148128"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-12 01:11:39.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-19 05:03:52.000000000","tz":600},"subject":"gerrit/acls : add NO_CODE_CHANGE","message":"gerrit/acls : add NO_CODE_CHANGE\n\nThe old copyAllScoresIfNoChange flag defaults to true, so any ACLs\nwithout it explicitly set to \"false\" (i.e. all of ours) have had the\nequivalant copy-condition flag \"changekind:NO_CODE_CHANGE\" added\nduring migration (see [1]).\n\nAny ACL\u0027s that were not specifying changekind:NO_CODE_CHANGE are\nupdated so that we are in sync with what is committed in Gerrit.\n\nWe know that updates that match changekind:TRIVIAL_REBASE also match\nchangekind:NO_CODE_CHANGE; however the migration has still added the\nlatter to all rules.  Thus to avoid confusion, we also add it here.\n\n[1] https://bugs.chromium.org/p/gerrit/issues/detail?id\u003d16839\n\nChange-Id: I1f11c07e3786bd1a68b43d908d939fde42ddb99c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66c99e87df21802bd1239312785dbe53f5794e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66c99e87df21802bd1239312785dbe53f5794e9f"}]},"branch":"refs/heads/master"},"6f70d8914590a290adf0c3947a96940fb6cb1b66":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-20 00:56:54.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/15/880115/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/880115/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/880115/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/880115/3"}}},"commit":{"parents":[{"commit":"e1f1945675800dbf888f06cefb53b07e7f92e694","subject":"Merge \"Add Dell Storage App to StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1f1945675800dbf888f06cefb53b07e7f92e694"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-12 01:11:39.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-20 00:53:16.000000000","tz":600},"subject":"gerrit/acls : add NO_CODE_CHANGE","message":"gerrit/acls : add NO_CODE_CHANGE\n\nThe old copyAllScoresIfNoChange flag defaults to true, so any ACLs\nwithout it explicitly set to \"false\" (i.e. all of ours) have had the\nequivalant copy-condition flag \"changekind:NO_CODE_CHANGE\" added\nduring migration (see [1]).\n\nAny ACL\u0027s that were not specifying changekind:NO_CODE_CHANGE are\nupdated so that we are in sync with what is committed in Gerrit.\n\nWe know that updates that match changekind:TRIVIAL_REBASE also match\nchangekind:NO_CODE_CHANGE; however the migration has still added the\nlatter to all rules.  Thus to avoid confusion, we also add it here.\n\n[1] https://bugs.chromium.org/p/gerrit/issues/detail?id\u003d16839\n\nChange-Id: I1f11c07e3786bd1a68b43d908d939fde42ddb99c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f70d8914590a290adf0c3947a96940fb6cb1b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f70d8914590a290adf0c3947a96940fb6cb1b66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
