)]}'
{"id":"openstack%2Fproject-config~880792","triplet_id":"openstack%2Fproject-config~master~I83583afb05e22d92002d92792f2bdf51507e06f2","project":"openstack/project-config","branch":"master","topic":"nocodechange","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-04-20 01:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I83583afb05e22d92002d92792f2bdf51507e06f2","subject":"acl : remove NO_CODE_CHANGE from Allow-Post-Review","status":"MERGED","created":"2023-04-19 05:05:33.000000000","updated":"2023-04-20 02:48:25.000000000","submitted":"2023-04-20 02:01:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880792-nocodechange","meta_rev_id":"974fdbd5b9cae7307987bf5da064ba98232ec0c6","_number":880792,"virtual_id_number":880792,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 02:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-04-20 00:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-04-20 00:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 01:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 05:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 09:11:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2023-04-19 23:09:00.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"2ee6d554bcb38e5b2170c98272b47697e4e82d6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-19 05:05:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3ad011f36e6ef67e082a4bf4958cd2bb3a5aa8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 05:14:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2f86d19bbd54adfa7d702759263e5c6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/27853d5d26bc4b269717d897ba73a644 : SUCCESS in 4m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8116c5e1b43c4395888c84bacdb1e409 : SUCCESS in 5m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/caf98b8174e84e989f3315b2eab153ca : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6184508c1324ab019ae029c1f2a8874d8db14f4b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-04-19 09:11:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfb5f082b27bb8d7eaffe9a4dae601f5bcd02204","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-04-19 23:09:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3643d01f4b5389ce15ef88bc96a6f44f2dcb7f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-20 00:56:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b179d59aa571f03e791535375c3354f15880e609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 01:05:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efb67926c9144d78bc9b5418d2b69511\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6cb3858731394ecea172263b13fd5cfe : SUCCESS in 5m 38s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f155e1cfd0de4efa826ba20cf8916561 : SUCCESS in 7m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/92223203b648476ea67a3814c35c069c : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7e93dcae9b638afc5b039236653be13f40048287","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-04-20 01:52:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"21c45a55e86c542dbbcdfb1fdf61160eb2267b49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 01:53:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"230fe0136f04ab0d031188a7cd3d5345b215f71f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 02:01:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b29d982a054fdcbeecf8a54213557f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cd6ba963567f443291d6992485309eff : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/060e46f555244dd2b0d0f199284c1d0c : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5fc4e773788412c2ae9899f9e929541df8daca00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 02:01:43.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"974fdbd5b9cae7307987bf5da064ba98232ec0c6","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 02:48:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/561a779593774eb0a531107f152d91ec\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b745238cae384027808eb2326b883b5d : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89aabb8c671ecca65c27a57ca4ed62ca49525f15","revisions":{"efb29264805cbaeb915fc1ffdc46519daf0ce9d1":{"kind":"REWORK","_number":1,"created":"2023-04-19 05:05:33.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/92/880792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/880792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/880792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/880792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/880792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/880792/1"}}},"commit":{"parents":[{"commit":"66c99e87df21802bd1239312785dbe53f5794e9f","subject":"gerrit/acls : add NO_CODE_CHANGE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66c99e87df21802bd1239312785dbe53f5794e9f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-19 04:47:58.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-19 05:03:53.000000000","tz":600},"subject":"acl : remove NO_CODE_CHANGE from Allow-Post-Review","message":"acl : remove NO_CODE_CHANGE from Allow-Post-Review\n\nThis is a special label that triggers jobs in a post-review pipeline\nthat has access to secrets.  Core members can trigger this to validate\na change that needs access to those secrets before merge.\n\nThis has been running with the NO_CODE_CHANGE copy-condition since\nit\u0027s inception since that is the default state; it was just added\nalong with all others explicity in prior change\nI1f11c07e3786bd1a68b43d908d939fde42ddb99c.\n\nHowever, it\u0027s probably best this label specify no copy conditions, so\nany modification at all resets it.  This removes it.\n\nChange-Id: I83583afb05e22d92002d92792f2bdf51507e06f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efb29264805cbaeb915fc1ffdc46519daf0ce9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efb29264805cbaeb915fc1ffdc46519daf0ce9d1"}]},"branch":"refs/heads/master"},"89aabb8c671ecca65c27a57ca4ed62ca49525f15":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-20 00:56:54.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/92/880792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/880792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/880792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/880792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/880792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/880792/2"}}},"commit":{"parents":[{"commit":"6f70d8914590a290adf0c3947a96940fb6cb1b66","subject":"gerrit/acls : add NO_CODE_CHANGE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f70d8914590a290adf0c3947a96940fb6cb1b66"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-19 04:47:58.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-04-20 00:53:19.000000000","tz":600},"subject":"acl : remove NO_CODE_CHANGE from Allow-Post-Review","message":"acl : remove NO_CODE_CHANGE from Allow-Post-Review\n\nThis is a special label that triggers jobs in a post-review pipeline\nthat has access to secrets.  Core members can trigger this to validate\na change that needs access to those secrets before merge.\n\nThis has been running with the NO_CODE_CHANGE copy-condition since\nit\u0027s inception since that is the default state; it was just added\nalong with all others explicity in prior change\nI1f11c07e3786bd1a68b43d908d939fde42ddb99c.\n\nHowever, it\u0027s probably best this label specify no copy conditions, so\nany modification at all resets it.  This removes it.\n\nChange-Id: I83583afb05e22d92002d92792f2bdf51507e06f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/89aabb8c671ecca65c27a57ca4ed62ca49525f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/89aabb8c671ecca65c27a57ca4ed62ca49525f15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
