)]}'
{"id":"openstack%2Fproject-config~880933","triplet_id":"openstack%2Fproject-config~master~Id484d75e3c742f1302dc92e23a3eb9c8a328c55c","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-04-20 11:45:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id484d75e3c742f1302dc92e23a3eb9c8a328c55c","subject":"Add sdk and cli xxx-service-core group","status":"MERGED","created":"2023-04-20 09:55:20.000000000","updated":"2023-04-20 11:51:20.000000000","submitted":"2023-04-20 11:45:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880933","meta_rev_id":"0c3f54d332992c994a96fcacb70f2423bb5c5552","_number":880933,"virtual_id_number":880933,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 11:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-20 11:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 11:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 10:08:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 11:37:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"45c18538dd57c1302ccca2d72c7328cc503ba408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-04-20 09:55:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86e898620b27d5784c1d27d523ad97296102b7f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:08:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fade127fdd4f3c8557e1751fb38ed9\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b9c1b840cef34fc9949c8403f4d8d18e : SUCCESS in 3m 04s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0344e1b57b6e4cb8867668d3cc4fedaa : SUCCESS in 6m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f06372dc0d994453a965f31c0650d68e : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"200b3c33004f4cb11a9d4e2f71a107b5b6205b02","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-04-20 11:37:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08203c33bd76dba47e9cfe14a4ae1fe161a98d2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:37:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab04987e2f1ffd261d5bf14156818543129a6f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:45:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ffa1338a86a48c8905eb822b4409ee9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/405387c17e524de5bfcabd09aefef852 : SUCCESS in 5m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/77ed5ebb5ee94d4397cfa789b2470973 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cddd7a8d1f6b57b2af8a2df57e473a4d6c2dd429","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:45:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c3f54d332992c994a96fcacb70f2423bb5c5552","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:51:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef73711128b64c55bb4d4bbce72f82e2\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1e3d74149f2941049ce17c82cd86d8d5 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a02e1ff8b4f70272548d6bd4a3b55d67ab888c6b","revisions":{"a02e1ff8b4f70272548d6bd4a3b55d67ab888c6b":{"kind":"REWORK","_number":1,"created":"2023-04-20 09:55:20.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/33/880933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/880933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/880933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/880933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/880933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/880933/1"}}},"commit":{"parents":[{"commit":"4b1e74a02e9d36d0bf93feb561b106c6f852cbdc","subject":"Merge \"acl : remove NO_CODE_CHANGE from Allow-Post-Review\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b1e74a02e9d36d0bf93feb561b106c6f852cbdc"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-04-20 09:40:42.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2023-04-20 09:40:42.000000000","tz":120},"subject":"Add sdk and cli xxx-service-core group","message":"Add sdk and cli xxx-service-core group\n\nIn order to have openstack service representatives having +2 permissions\non sdk and cli to signal the change is ok from their side while still\ncontrolling their ability to gate changes create 2 new groups with only\n+2 vote right. This will help to give services wider ability to review\nand approve changes for their services while still keeping final word on\nthe sdk/cli cores side and thus avoid privileges misuse.\n\nChange-Id: Id484d75e3c742f1302dc92e23a3eb9c8a328c55c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a02e1ff8b4f70272548d6bd4a3b55d67ab888c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a02e1ff8b4f70272548d6bd4a3b55d67ab888c6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
