)]}'
{"id":"openstack%2Fproject-config~882075","triplet_id":"openstack%2Fproject-config~master~I73a07a3542ca26ddfcf01eab1d7be11cec70da85","project":"openstack/project-config","branch":"master","topic":"normalize-acls","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-05-02 19:24:12.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-05-10 15:24:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73a07a3542ca26ddfcf01eab1d7be11cec70da85","subject":"Add an \"apply\" transformation which applies all","status":"MERGED","created":"2023-05-02 15:25:46.000000000","updated":"2023-05-10 15:24:58.000000000","submitted":"2023-05-10 15:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"882075-normalize-acls","meta_rev_id":"8abfd3dcc6c6a2d51362e8a8cdf1a92fa2ee4002","_number":882075,"virtual_id_number":882075,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 15:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-05-02 19:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-02 23:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-05-10 15:10:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 15:32:37.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-05-02 15:37:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 23:17:35.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"265ce5171092b61a9316a3405c23490c24affa96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 15:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29417c8d58859d1e1ed426bb9b6d9e97b1a4cd5b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-02 15:32:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa5d04e6505aa38b279e3ff66383fb2d438847e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 15:37:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0979f57c89d7497d96508a78747142eb\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8a153ef03e1a4ed58dd161b9ac181968 : SUCCESS in 2m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a45b59d9b9ba4f7e9624cf6f1bdfd355 : SUCCESS in 7m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aef1a28e99f8443389067d58c2ce8cf5 : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"250b14854b83346ec0ba09d88995b3d9b18122c3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 15:40:30.000000000","message":"Set Work In Progress\n\nHold on, I may be misreading the comments.","accounts_in_message":[],"_revision_number":1},{"id":"f4a3e7f518fba851e66bf53eb93ad70a03296416","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-02 15:54:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"789744cbf59f531b296b15c5af199278f071a4ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 16:02:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861345e0b859487080f622b32876fb84\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8d3882bec974455a8089f67851d030be : SUCCESS in 3m 34s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/15e2454c6cb6445ca0fbd20283ce6425 : SUCCESS in 6m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3bb090d67de343539dc9ab76fe24cfd0 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8cbec61e2c54f9296b6cf47b38bff0ac434103e6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-02 19:24:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3337446ef2dbf005167e8b1a82a9eb02afa97f6a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-05-02 23:17:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f38e1bf21ad2875d0dd2890a6f0db44bed8b187b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-03 12:28:21.000000000","message":"Patch Set 2:\n\nI have a feeling (based on the whitespace issues I see in proposed changes) that at least some people proposing these changes are doing it from Windows, so if we\u0027re going to recommend an invocation tersely in the job output we may be better off referring to documentation (Infra Manual or elsewhere) that can go into more detail in a cross-platform-friendly manner.\n\nI guess another option would be to add a tox/nox env specifically for running this.","accounts_in_message":[],"_revision_number":2},{"id":"ea1dd8ec975ab90126ae8fc891e06cde733b1393","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-08 20:32:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c412b13032b31d69398ec90f217b91387738aa7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-09 19:36:49.000000000","message":"Patch Set 2:\n\nCorresponding tox testenv added in followup I5f0903849ea72f9d1c81164bed4c651c9423f1fc so we can bikeshed the implementation for it separately.","accounts_in_message":[],"_revision_number":2},{"id":"65cc23baf5f1566177f547436700596630495b3f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-10 15:10:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81214438bfe7be7b57757d7f15513db367a1cd03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:11:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"013338e4c0e0ff1608fd37855de2c294fbfc0e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:24:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c843765108646208da468862aa30296\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4a2bdd0c89245d782f9fe89aa79d3fe : SUCCESS in 6m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7a0ec38dd0d845e682e9672cfc5939d0 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8abfd3dcc6c6a2d51362e8a8cdf1a92fa2ee4002","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:24:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f913313781a050206ee2b320ee17b1fc5d1f29ef","revisions":{"9870aee44fbfc6cbebbc993c2a4d0b15f6e59c44":{"kind":"REWORK","_number":1,"created":"2023-05-02 15:25:46.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/75/882075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/882075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/882075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/882075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/882075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/882075/1"}}},"commit":{"parents":[{"commit":"9329bd6a787ee94b5a27e51c67037d69b54f145b","subject":"Merge \"Retire puppet-rally - Step 1: End project Gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9329bd6a787ee94b5a27e51c67037d69b54f145b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-02 15:23:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-02 15:23:15.000000000","tz":0},"subject":"Fix the \"all\" transformation so it actually works","message":"Fix the \"all\" transformation so it actually works\n\nThe Gerrit ACL normalization script incorrectly included its dry run\ntransformation (0) in the generated range when the \"all\"\ntransformation keyword was supplied. Correctly exclude 0 from the\nrange so that you can actually use that shortcut in order to\ndirectly normalize ACLs instead of merely printing the result to the\nterminal.\n\nChange-Id: I73a07a3542ca26ddfcf01eab1d7be11cec70da85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9870aee44fbfc6cbebbc993c2a4d0b15f6e59c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9870aee44fbfc6cbebbc993c2a4d0b15f6e59c44"}]},"branch":"refs/heads/master"},"f913313781a050206ee2b320ee17b1fc5d1f29ef":{"kind":"REWORK","_number":2,"created":"2023-05-02 15:54:07.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/75/882075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/882075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/882075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/882075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/882075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/882075/2"}}},"commit":{"parents":[{"commit":"9329bd6a787ee94b5a27e51c67037d69b54f145b","subject":"Merge \"Retire puppet-rally - Step 1: End project Gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9329bd6a787ee94b5a27e51c67037d69b54f145b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-02 15:23:15.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-02 15:48:44.000000000","tz":0},"subject":"Add an \"apply\" transformation which applies all","message":"Add an \"apply\" transformation which applies all\n\nThe Gerrit ACL normalization script has an \"all\" transformation\nwhich reports all transformation results to the screen. Add a\nsimilar \"apply\" shortcut to directly transform the file being\noperated on in a non dry run fashion.\n\nChange-Id: I73a07a3542ca26ddfcf01eab1d7be11cec70da85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f913313781a050206ee2b320ee17b1fc5d1f29ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f913313781a050206ee2b320ee17b1fc5d1f29ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
