)]}'
{"id":"openstack%2Fproject-config~882676","triplet_id":"openstack%2Fproject-config~master~I86370d45b017bb0b4667055a3e737c383b851776","project":"openstack/project-config","branch":"master","topic":"fix-acls","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-05-09 12:45:09.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-05-09 12:24:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I86370d45b017bb0b4667055a3e737c383b851776","subject":"Adjust ACL for ovn-bgp-agent releases","status":"MERGED","created":"2023-05-09 08:24:32.000000000","updated":"2023-05-09 12:50:42.000000000","submitted":"2023-05-09 12:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882676-fix-acls","meta_rev_id":"8d860200672c0120fdcfbb1169043de47d1e55cf","_number":882676,"virtual_id_number":882676,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-09 12:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-09 12:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-09 12:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 08:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-09 12:18:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"5780a35ca2b6aa232498ffa8589aa8d0946e174c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-05-09 08:24:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb51ed68229447d77f5ac15db6894aa1c73d68e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 08:31:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de1d017690ca40649693f17b3086929e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1469ef022eec484aa06789f78d5526c8 : SUCCESS in 6m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b49fba9f52ef43d2b32707bd56993321 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3c8394210e94c546b32bbf9779687386083cab4a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-09 12:18:06.000000000","message":"Patch Set 1: Workflow-1\n\nFor this to work, it\u0027ll also need access.inheritFrom\u003dopenstack/meta-config to pick up the release management permissions. I\u0027ll push up a revision with that and approve.","accounts_in_message":[],"_revision_number":1},{"id":"686efb9c1daa0277a46dcf3c3c5e8f9b56a3b911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-09 12:21:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dc6482f5962a89e9066d73285b084893349c6953","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-09 12:24:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI also gave the core reviewers toggleWipState so they don\u0027t come asking admins to un-WIP changes when the change owner is on holiday.","accounts_in_message":[],"_revision_number":2},{"id":"9bdb72383b672ba553cba0f8cc6a6014765ffe2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:36:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85df31210e994d7bae7455f9ea01e35c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d8fc2052ef7345b5a883941473718b9b : SUCCESS in 7m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3f1d9d0d342a44f18fe904a72bd4eb1d : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cebf1174d2be3d7d199b90de9b9ef69e9b9c96e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:36:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36bd5709dfdf4b0e441ebd78042da062dad6955d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:45:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d2c53dc17f4d0d8a07e50b97732dfc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18b174e9a1bf40219c53fa07963b19b5 : SUCCESS in 7m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/08ef8888292c463d8ddb3c54d5446a8a : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e42b5c1f45cb197d3d4653f2f72e5a40e5baadee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:45:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8d860200672c0120fdcfbb1169043de47d1e55cf","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:50:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013bd6faba4c471693bff1e9d10b67d5\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/58d66cf0f715413292895f63f8d2ab60 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"534a381ac18ba36627c05397b22ba8762354dcc4","revisions":{"b141ceeff1c79f85d79561436d0d8d281d28a86d":{"kind":"REWORK","_number":1,"created":"2023-05-09 08:24:32.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/76/882676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/882676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/882676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/882676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/882676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/882676/1"}}},"commit":{"parents":[{"commit":"16fc79b3879ad31acaf2038bcaec03b5b06cda19","subject":"Merge \"Retire puppet-rally - Step 5: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16fc79b3879ad31acaf2038bcaec03b5b06cda19"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-05-09 08:21:01.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-05-09 08:21:01.000000000","tz":120},"subject":"Adjust ACL for ovn-bgp-agent releases","message":"Adjust ACL for ovn-bgp-agent releases\n\novn-bgp-agent was recently moved from x/ to openstack/ under the Neutron\nteam. It should from now on be tagged through Release management tooling\nrather than directly, so we adjust the Gerrit ACL accordingly.\n\nChange-Id: I86370d45b017bb0b4667055a3e737c383b851776\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b141ceeff1c79f85d79561436d0d8d281d28a86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b141ceeff1c79f85d79561436d0d8d281d28a86d"}]},"branch":"refs/heads/master"},"534a381ac18ba36627c05397b22ba8762354dcc4":{"kind":"REWORK","_number":2,"created":"2023-05-09 12:21:39.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/76/882676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/882676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/882676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/882676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/882676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/882676/2"}}},"commit":{"parents":[{"commit":"16fc79b3879ad31acaf2038bcaec03b5b06cda19","subject":"Merge \"Retire puppet-rally - Step 5: Remove Project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16fc79b3879ad31acaf2038bcaec03b5b06cda19"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2023-05-09 08:21:01.000000000","tz":120},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-09 12:21:28.000000000","tz":0},"subject":"Adjust ACL for ovn-bgp-agent releases","message":"Adjust ACL for ovn-bgp-agent releases\n\novn-bgp-agent was recently moved from x/ to openstack/ under the Neutron\nteam. It should from now on be tagged through Release management tooling\nrather than directly, so we adjust the Gerrit ACL accordingly.\n\nChange-Id: I86370d45b017bb0b4667055a3e737c383b851776\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/534a381ac18ba36627c05397b22ba8762354dcc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/534a381ac18ba36627c05397b22ba8762354dcc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
