)]}'
{"id":"openstack%2Fproject-config~882750","triplet_id":"openstack%2Fproject-config~master~I5f0903849ea72f9d1c81164bed4c651c9423f1fc","project":"openstack/project-config","branch":"master","topic":"normalize-acls","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-05-10 15:10:10.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-05-10 15:25:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f0903849ea72f9d1c81164bed4c651c9423f1fc","subject":"Add a convenient normalize-gerrit-acl tox testenv","status":"MERGED","created":"2023-05-09 19:36:10.000000000","updated":"2023-05-10 15:26:23.000000000","submitted":"2023-05-10 15:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882750-normalize-acls","meta_rev_id":"837889c0a52fbbd0a81ea056d606dc62143d76ef","_number":882750,"virtual_id_number":882750,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 15:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-05-10 15:10:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-10 01:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-05-10 15:10:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 19:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-09 23:15:47.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-05-10 01:46:11.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"2d0ea3fd7a3fa8356f82e0bd53111bba012432fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-09 19:36:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a62df446ef4573d24dba4ebb4e4caa6b6f4d6abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 19:47:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9655b45c4fb34e7da4d4cd857b3a9326\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/5a8ed68b375749ccbf4810d519eebab6 : SUCCESS in 2m 53s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8f1277a7598a49ab8d49ec5d11605737 : SUCCESS in 6m 06s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/2cb229d8c33140009bcf9aa0018e29d5 : SUCCESS in 3m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/125f84bc94044c5aa34e34138d1d0c48 : SUCCESS in 3m 17s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/9c92614086c148469b40a9a8ac6c224c : SUCCESS in 3m 00s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f655f973fa4b4537ba6dde67f8ecc40b : SUCCESS in 5m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8781bc4aa40e44afb045ca461b503889 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d7732056bbe9383e0f9e286b67f3488190893059","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-09 23:15:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"578b871bde5890d4c58275055139ce8dfd4435c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-10 00:29:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"632e48b21c7a85fc88876f2c267697ab7fde1cc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 00:41:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5dcd17471714c33b8d57df0a9e362bc\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/2be794087cc840978611907bb4ae3c9f : SUCCESS in 2m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/16e5bd1d37d9444c9897bc3ed0932893 : SUCCESS in 6m 21s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/286d5a09f5144987aa6827d5cc450260 : SUCCESS in 3m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9fee6deffb024aaea71b06ae536717ad : SUCCESS in 3m 25s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/04b20f57a4ee49818ce64637fdf8224c : SUCCESS in 3m 05s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/9e85bb461b594e4cab6cb199559e9b0c : SUCCESS in 5m 44s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/285ac02b54e54ec5b307114162cb339c : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"12398949d1fd0e826f0b82cdab278f41e97e5114","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-05-10 01:46:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d110e4d38dc2fe00f2fb2660363fc82c9907de31","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-10 15:10:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba796e2f5c48d27b24af079047280c5da29154c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:11:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90d26abbc6dc09566463391e46387206a0a55c11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:25:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0e337ae8dc4767a42c8bcb9752547d\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/35dabfa80f824f0682ff1a9e8e142833 : SUCCESS in 4m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/55e08bfbb3c3467b950f9dbcf4700f1b : SUCCESS in 7m 12s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/7fe2d162512e4c808a20aedb32c4a96f : SUCCESS in 2m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/218acf4b59fc4496b1e4593181a0af21 : SUCCESS in 3m 36s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/706da203436c49cb9441df36ad381482 : SUCCESS in 3m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/255624614f9946b3b7f0d74f7e36ac5c : SUCCESS in 5m 44s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/23c1828fdd704d0c84a4122c6479610a : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1a82e4d5af70a66dbcd1f10828e6abd110c1b011","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:25:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"837889c0a52fbbd0a81ea056d606dc62143d76ef","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:26:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b38e0ddc9b446208abf2e28608c5118\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/79871dba13fe4538903f3cb6bcce0928 : SUCCESS in 45s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/a347de8ecfd940cba6c3e0667cd42bfa : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1832dcdad7807fe083ff75c26b0efa0349f31120","revisions":{"a4c131ca921f3bc435dd5d0425d23a280a4307e9":{"kind":"REWORK","_number":1,"created":"2023-05-09 19:36:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/50/882750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/882750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/882750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/882750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/882750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/882750/1"}}},"commit":{"parents":[{"commit":"f913313781a050206ee2b320ee17b1fc5d1f29ef","subject":"Add an \"apply\" transformation which applies all","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f913313781a050206ee2b320ee17b1fc5d1f29ef"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-09 19:32:50.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-09 19:34:46.000000000","tz":0},"subject":"Add a convenient normalize-gerrit-acl tox testenv","message":"Add a convenient normalize-gerrit-acl tox testenv\n\nUsers struggle with Gerrit ACL formatting, often going through 4 or\n5 revisions to get things like indents, trailing whitespace,\nalphabetical ordering and final newlines correct. With this testenv,\nusers can run something like this to auto-reformat their Gerrit\nACLs...\n\n  tox -e normalize-gerrit-acl -- gerrit/acls/foo/bar.config\n\nWe\u0027ll include this as a suggestion in the Infra Manual section on\nadding projects as well.\n\nChange-Id: I5f0903849ea72f9d1c81164bed4c651c9423f1fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4c131ca921f3bc435dd5d0425d23a280a4307e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4c131ca921f3bc435dd5d0425d23a280a4307e9"}]},"branch":"refs/heads/master"},"1832dcdad7807fe083ff75c26b0efa0349f31120":{"kind":"REWORK","_number":2,"created":"2023-05-10 00:29:44.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/50/882750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/882750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/882750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/882750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/882750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/882750/2"}}},"commit":{"parents":[{"commit":"f913313781a050206ee2b320ee17b1fc5d1f29ef","subject":"Add an \"apply\" transformation which applies all","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f913313781a050206ee2b320ee17b1fc5d1f29ef"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-09 19:32:50.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-10 00:29:25.000000000","tz":0},"subject":"Add a convenient normalize-gerrit-acl tox testenv","message":"Add a convenient normalize-gerrit-acl tox testenv\n\nUsers struggle with Gerrit ACL formatting, often going through 4 or\n5 revisions to get things like indents, trailing whitespace,\nalphabetical ordering and final newlines correct. With this testenv,\nusers can run something like this to auto-reformat their Gerrit\nACLs...\n\n  tox -e normalize-gerrit-acl -- gerrit/acls/foo/bar.config\n\nWe\u0027ll include this as a suggestion in the Infra Manual section on\nadding projects as well.\n\nChange-Id: I5f0903849ea72f9d1c81164bed4c651c9423f1fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1832dcdad7807fe083ff75c26b0efa0349f31120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1832dcdad7807fe083ff75c26b0efa0349f31120"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
