)]}'
{"id":"openstack%2Fproject-config~883303","triplet_id":"openstack%2Fproject-config~master~I68e06bbfaf0b38582320f09659dfc462016eacf8","project":"openstack/project-config","branch":"master","topic":"ansible-lint","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-05-16 21:04:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68e06bbfaf0b38582320f09659dfc462016eacf8","subject":"Pin ansible-compat\u003c4 until we upgrade ansible-lint","status":"MERGED","created":"2023-05-16 20:07:50.000000000","updated":"2023-05-16 21:05:51.000000000","submitted":"2023-05-16 21:04:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883303-ansible-lint","meta_rev_id":"dc8e3ca7fc8c2b7d3fe38f763ed1e4be95016a64","_number":883303,"virtual_id_number":883303,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-16 21:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-05-16 20:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-16 20:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-05-16 20:53:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 20:11:16.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2023-05-16 20:19:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 20:53:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"a07d5e151fef854cda46dda9b524103ff8ca9ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-16 20:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e37762ebb33ca71731a29cb6239b0471d0cc4a97","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-05-16 20:11:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d744c5518588d3e4440c9befa36ca5ca0f96767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 20:19:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe1146adfeb4246a7bd428993c74ba9\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/565178f8e3a84fdcb25be6b67440d1fe : SUCCESS in 4m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4da6c7dba2534805b775d027bde915af : SUCCESS in 5m 55s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/c2ab599964954394a4ac904af017a649 : SUCCESS in 4m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a0bfc7fac95842948e8735b2da902b12 : SUCCESS in 3m 09s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/d5b3a0b1a44c4de0bbb9067b2321c21a : SUCCESS in 3m 51s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/00daddd2fa9445c7a77cd089ddd2da8d : SUCCESS in 5m 37s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/722e3feec6504bcbb8cf28856b9f0b21 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"822e1e05ac2455d7720fcde427cc30509f85223c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-16 20:53:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99d77262fb1615e7d0571f9e01ccf9a76c8140b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 20:54:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c24ac7d07b67e51916698837c16866726b6e1ecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 21:04:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d59a04ece9724c7a814d17422294858c\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/4638c425afce4bb5962a8f1998759ee1 : SUCCESS in 3m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6a1a60090fc442cab47eba1bdffcca23 : SUCCESS in 5m 58s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/349aa9c8aa8a4f758b905478371d04d4 : SUCCESS in 4m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/04f3d58342984388af13b56c8bf75aa4 : SUCCESS in 3m 38s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/1d4c0d32c74042dba650bfbe89d2a499 : SUCCESS in 3m 07s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/2be8d1d38ae541a1b3ec348ed7f1b0e8 : SUCCESS in 6m 10s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/aa533d4fe2264fd18364fef5e0b1bbe4 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4ca92677370a5e4ac911095f073dcf1c26d49e39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 21:04:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc8e3ca7fc8c2b7d3fe38f763ed1e4be95016a64","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 21:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5440ee18e940a98fa7421934d20618\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/587a2a4a848e4feb9cef4e909ddae51a : SUCCESS in 37s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/956e8d85de15486da29e97dd230a6066 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f87307b86a33a3a8f2aa39b9096724f1542d27f8","revisions":{"f87307b86a33a3a8f2aa39b9096724f1542d27f8":{"kind":"REWORK","_number":1,"created":"2023-05-16 20:07:50.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/03/883303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/883303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/883303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/883303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/883303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/883303/1"}}},"commit":{"parents":[{"commit":"c609969fa657a3bef9d916f8d7525404852aa166","subject":"Merge \"Add nebulous/component-template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c609969fa657a3bef9d916f8d7525404852aa166"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-16 20:05:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-05-16 20:05:54.000000000","tz":0},"subject":"Pin ansible-compat\u003c4 until we upgrade ansible-lint","message":"Pin ansible-compat\u003c4 until we upgrade ansible-lint\n\nThere was a regression introduced in ansible-compat v4 which stopped\nansible-lint from working. This was fixed in ansible-lint 6.16.0 but\nwe\u0027ve pinned it for so long now that there are a lot of rule\nviolations to fix or mark for skipping. In the interim, pin back all\nnew versions of ansible-compat until someone has time to fix things\nor rip out ansible-lint entirely.\n\nChange-Id: I68e06bbfaf0b38582320f09659dfc462016eacf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f87307b86a33a3a8f2aa39b9096724f1542d27f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f87307b86a33a3a8f2aa39b9096724f1542d27f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
