)]}'
{"id":"openstack%2Fproject-config~883523","triplet_id":"openstack%2Fproject-config~master~I9c7e063bfe7d7324c6833560ad4b1264752f62fb","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2023-05-18 15:32:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9c7e063bfe7d7324c6833560ad4b1264752f62fb","subject":"Fix voting groups of starlingx/app-harbor","status":"MERGED","created":"2023-05-18 15:05:32.000000000","updated":"2023-05-18 15:38:36.000000000","submitted":"2023-05-18 15:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883523","meta_rev_id":"96ec2edd902e8d6365891c340512f6f6ab0d1266","_number":883523,"virtual_id_number":883523,"owner":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-18 15:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-18 15:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-18 15:05:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 15:05:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-05-18 15:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78a648de6d5c4730cfdc292adb4b4a157e46b376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2023-05-18 15:05:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17e91f9d7154efcccf585610e182bd8490810698","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-18 15:05:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e858c4dd8c3d136e2e232fbe776242483283eb27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:21:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d00c748ca34cbd979cfaf82341fc24\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/23ba891d344c479f9cca16ae2d06f9cb : SUCCESS in 8m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7b60e093c9ee46c6b1340b4b9df61960 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"168ebc6aaab3cecc4b7b27410d12667f8bff6243","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:22:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b0d30808ce286baff9bef699506331814d6a086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:32:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/047b4058ec9341b3aeb4e47b204dd6ee\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/66c80a2901034be1af3aedae30674a00 : SUCCESS in 5m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f5ac554c0b224d63a7820823ba65d786 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ffe2503cd39b78108c961aa5d4d8bd8e2a29cf55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:32:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"96ec2edd902e8d6365891c340512f6f6ab0d1266","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87bdf4f7063247128bc574653520c810\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/69d41ba34f844e9db93fb33bb3d6a917 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"628b5ba4d60b4c9fa53d4a10c3adbe0818805734","revisions":{"628b5ba4d60b4c9fa53d4a10c3adbe0818805734":{"kind":"REWORK","_number":1,"created":"2023-05-18 15:05:32.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/23/883523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/883523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/883523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/883523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/883523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/883523/1"}}},"commit":{"parents":[{"commit":"bc8b74f30108d1ae3355ce0ecaef7dd64db0bde9","subject":"Fix the NebulOuS tenant config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc8b74f30108d1ae3355ce0ecaef7dd64db0bde9"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2023-05-18 15:05:24.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2023-05-18 15:05:24.000000000","tz":-240},"subject":"Fix voting groups of starlingx/app-harbor","message":"Fix voting groups of starlingx/app-harbor\n\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\nChange-Id: I9c7e063bfe7d7324c6833560ad4b1264752f62fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/628b5ba4d60b4c9fa53d4a10c3adbe0818805734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/628b5ba4d60b4c9fa53d4a10c3adbe0818805734"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
