)]}'
{"id":"openstack%2Fproject-config~884563","triplet_id":"openstack%2Fproject-config~master~Id7290012f88881aec0bc56ff1a3ae879b309d37e","project":"openstack/project-config","branch":"master","topic":"zuul-config-errors","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2023-06-01 09:23:50.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-06-01 09:15:39.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Id7290012f88881aec0bc56ff1a3ae879b309d37e","subject":"Stop testing github hosted projects in openstack tenant","status":"MERGED","created":"2023-05-27 07:54:45.000000000","updated":"2023-06-01 09:23:50.000000000","submitted":"2023-06-01 09:23:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"884563-zuul-config-errors","meta_rev_id":"9303bcdc19d0221e99b5c3b6b521c2bd15134956","_number":884563,"virtual_id_number":884563,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 09:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2023-05-31 17:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2023-06-01 09:15:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-27 07:55:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-27 07:55:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 18:50:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-05-30 19:23:45.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2023-05-31 17:04:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"548d5f73cc007b50c3747a6775761c794fa64938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-27 07:54:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5471827bcf0097e8948a425b296ed40da5adef43","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-27 07:55:03.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6609fd194580fd7d11469b974cea4222c78eb9ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 07:55:05.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered an error while accessing the repo ansible/ansible.\nThe error was:\n\n  Merge mode rebase not supported by project github.com/ansible/ansible","accounts_in_message":[],"_revision_number":1},{"id":"dbcbc80791ebf49d8364fac5ff1123de6dbd1f7f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-27 07:55:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5928856013d870df2c0a1e6df7f621d77fdbe718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-27 07:56:56.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77bffb9e52420ba67156aae6e679e1a1502869c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 07:57:19.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  not a valid value for dictionary value @ data[\u0027merge-mode\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: ansible/ansible\n      default-branch: devel\n      # Note: Specifying the merge-mode shouldn\u0027t be necessary since we\n      # do not gate this project, but zuul checks it anyway and generates\n      # a config error on mismatch\n      merge-mode: squash\n      third-party-check:\n        jobs:\n          - openstacksdk-ansible-stable-2.6-functional-devstack:\n              branches: stable-2.6\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack_inventory.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n          - shade-ansible-stable-2.5-functional-devstack:\n              branches: stable-2.5\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 24, column 3","accounts_in_message":[],"_revision_number":2},{"id":"cc583953a599ac8decb9281f5fdeab551690cc5d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-27 07:57:21.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":2},{"id":"285d9205328c55123d6bcdccd33c69c2d8aae4c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-27 07:59:52.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1b7fbdc0e02a198ebafc310eb106fcde7627aea","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-27 08:00:18.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":3},{"id":"c0a2694ed5422216994a38cbcc1973fb1c5334fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 08:00:25.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered an error while accessing the repo ansible/ansible.\nThe error was:\n\n  Merge mode squash-merge not supported by project github.com/ansible/ansible","accounts_in_message":[],"_revision_number":3},{"id":"7808ae9412b69383aa4c561aa9c4720be0be4ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-27 08:01:17.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"457bc780efdd19465ab58cf514e5dbd68339e4d7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-27 08:01:37.000000000","message":"Patch Set 4:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":4},{"id":"9c4bb9fe59c93a4850015e403fc29bfc880480d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 08:01:39.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered an error while accessing the repo pytest-dev/pytest-\ntestinfra.  The error was:\n\n  Merge mode rebase not supported by project github.com/pytest-dev/pytest-testinfra","accounts_in_message":[],"_revision_number":4},{"id":"bef07edf5cb52ec3a97a5db902c437496accf3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-27 08:02:16.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5469e23052aa9a9cf3eda40ec7e63d0b7bb36a79","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-27 08:02:35.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":5},{"id":"60df34a9ec7aaf3fb1a7d0cb5bc7441eecee7df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 08:26:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f563623dbeb4952a0b700c907aa4411\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ed45cceb9e0e4cd981dc1fd4fcddc63d : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":5},{"id":"8a5ea998e74fb28369ec5ab1106bef05fbb205ab","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-30 18:50:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"27b44ef668b3cea99ef8f2c62f6d4fb942637e91","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2023-05-30 19:23:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4e3c8560084a577616b400fc121602c02d08862e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-30 19:50:44.000000000","message":"Patch Set 5: -Code-Review\n\nAh, right Zuul needs to at least be able to faithfully emulate the future merged state of a PR in that repo regardless of whether it gates those PRs, so wants to try to match the merge mode configured in GitHub in order to do so.","accounts_in_message":[],"_revision_number":5},{"id":"e8cd4da7978abf271ad6145725ed736691b144c9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-31 05:21:38.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ec369494ade345a6474fa84bab9e0c5f88af7e68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-31 08:07:43.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"26379f5724e93a7d2f613c670da5b997c9057ca7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-31 08:08:02.000000000","message":"Patch Set 6:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Pipelines may not be defined in untrusted repos, they may only be\n  defined in config repos.\n\nThe error appears in the following pipeline stanza:\n\n  pipeline:\n      name: check\n      description: |\n        Newly uploaded patchsets enter this pipeline to receive an\n        initial +/-1 Verified vote.\n      success-message: Build succeeded (check pipeline).\n      failure-message: |\n        Build failed (check pipeline).  For information on how to proceed, see\n        https://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n      manager: independent\n      precedence: low\n      require:\n        gerrit:\n          open: True\n          current-patchset: True\n      trigger:\n        gerrit:\n          - event: patchset-created\n          - event: change-restored\n          - event: comment-added\n            comment: (?i)^(Patch Set [0-9]+:)?( [\\w\\\\+-]*)*(\\n\\n)?\\s*recheck\n          - event: comment-added\n            require-approval:\n              - Verified: [-1, -2]\n                username: zuul\n            approval:\n              - Workflow: 1\n        github:\n          - event: pull_request\n            action:\n              - opened\n              - changed\n              - reopened\n          - event: pull_request\n            action: comment\n            comment: (?i)^\\s*recheck\\s*$\n          - event: check_run\n      start:\n        github:\n          check: \u0027in_progress\u0027\n          comment: false\n      success:\n        gerrit:\n          # Note that gerrit keywords are case-sensitive.\n          Verified: 1\n        github:\n          check: \u0027success\u0027\n          comment: false\n      failure:\n        gerrit:\n          Verified: -1\n        github:\n          check: \u0027failure\u0027\n          comment: false\n\n  in \"openstack/project-config/zuul.d/pipelines.yaml@master\", line 4, column 3","accounts_in_message":[],"_revision_number":6},{"id":"93f12b4e74cbb35946d5b24d8472ab1c9427667c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 08:22:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfd53f900ba54dacb5232fd824db5d73\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fce209e1d8f447708eb9d7ca56349d2d : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":6},{"id":"61f265ee8ac32bb5518324889e7c1d7a1f38df76","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-31 17:04:08.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8973dc80e9865d980ccb1d87638580e9d996a65b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-01 09:15:39.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3ef8160f837720cd0e597c00ccb6f89fc7d1b610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 09:16:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cc2832922083be0df48e61485d1f0cb39af8548f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 09:23:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a558749fe9d8413aa9d78982cd8db585\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bda518697ce247d58f0c43951a8c6562 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":6},{"id":"9303bcdc19d0221e99b5c3b6b521c2bd15134956","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 09:23:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bdaba25653be6af71f1fd4dd49e1346e1c6536ba","revisions":{"19a52011a639769deb8bcbaceb4ae05da1ad8a20":{"kind":"REWORK","_number":1,"created":"2023-05-27 07:54:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/884563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/884563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/884563/1"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"subject":"Fix zuul config errors for github hosted projects","message":"Fix zuul config errors for github hosted projects\n\nThis should technically not be necessary, but zuul complains anyway, so\nclean these up.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id7290012f88881aec0bc56ff1a3ae879b309d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/19a52011a639769deb8bcbaceb4ae05da1ad8a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/19a52011a639769deb8bcbaceb4ae05da1ad8a20"}]},"branch":"refs/heads/master"},"54f5da6be0458ee1f40c01665d07e01cd48073e0":{"kind":"REWORK","_number":2,"created":"2023-05-27 07:56:56.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/884563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/884563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/884563/2"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-05-27 07:56:56.000000000","tz":0},"subject":"Fix zuul config errors for github hosted projects","message":"Fix zuul config errors for github hosted projects\n\nThis should technically not be necessary, but zuul complains anyway, so\nclean these up.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id7290012f88881aec0bc56ff1a3ae879b309d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/54f5da6be0458ee1f40c01665d07e01cd48073e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/54f5da6be0458ee1f40c01665d07e01cd48073e0"}]},"branch":"refs/heads/master"},"accdbe5cee3c0e76b746705cdb1448e7cfa1cbd5":{"kind":"REWORK","_number":3,"created":"2023-05-27 07:59:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/884563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/884563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/884563/3"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-05-27 07:59:52.000000000","tz":0},"subject":"Fix zuul config errors for github hosted projects","message":"Fix zuul config errors for github hosted projects\n\nThis should technically not be necessary, but zuul complains anyway, so\nclean these up.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id7290012f88881aec0bc56ff1a3ae879b309d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/accdbe5cee3c0e76b746705cdb1448e7cfa1cbd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/accdbe5cee3c0e76b746705cdb1448e7cfa1cbd5"}]},"branch":"refs/heads/master"},"e1acf1311c53f616f6d70688f0bc8d552c298a86":{"kind":"REWORK","_number":4,"created":"2023-05-27 08:01:17.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/884563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/884563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/884563/4"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-05-27 08:01:17.000000000","tz":0},"subject":"Fix zuul config errors for github hosted projects","message":"Fix zuul config errors for github hosted projects\n\nThis should technically not be necessary, but zuul complains anyway, so\nclean these up.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id7290012f88881aec0bc56ff1a3ae879b309d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1acf1311c53f616f6d70688f0bc8d552c298a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1acf1311c53f616f6d70688f0bc8d552c298a86"}]},"branch":"refs/heads/master"},"69800fce72f9bcfa834f6fd030395b4067e7ad80":{"kind":"REWORK","_number":5,"created":"2023-05-27 08:02:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/884563/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/884563/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/884563/5"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-05-27 08:02:16.000000000","tz":0},"subject":"Fix zuul config errors for github hosted projects","message":"Fix zuul config errors for github hosted projects\n\nThis should technically not be necessary, but zuul complains anyway, so\nclean these up.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id7290012f88881aec0bc56ff1a3ae879b309d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69800fce72f9bcfa834f6fd030395b4067e7ad80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69800fce72f9bcfa834f6fd030395b4067e7ad80"}]},"branch":"refs/heads/master"},"bdaba25653be6af71f1fd4dd49e1346e1c6536ba":{"kind":"REWORK","_number":6,"created":"2023-05-31 08:07:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/884563/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/884563/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/884563/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/884563/6"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-05-27 07:53:37.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-05-31 08:07:43.000000000","tz":0},"subject":"Stop testing github hosted projects in openstack tenant","message":"Stop testing github hosted projects in openstack tenant\n\nZuul currently has some issue when dealing with the merge mode for\nprojects hosted on github. Since due to the resulting zuul config\nerror no jobs are executed anyway, clean up the config to at least\nget rid of the errors.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Id7290012f88881aec0bc56ff1a3ae879b309d37e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bdaba25653be6af71f1fd4dd49e1346e1c6536ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bdaba25653be6af71f1fd4dd49e1346e1c6536ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
