)]}'
{"id":"openstack%2Fproject-config~884718","triplet_id":"openstack%2Fproject-config~master~Ib60099f79de902906b8d1956cc202aeb86587cb5","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-05-30 18:35:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib60099f79de902906b8d1956cc202aeb86587cb5","subject":"Remove retied projects from openstack-helm group","status":"MERGED","created":"2023-05-30 15:41:52.000000000","updated":"2023-05-30 21:09:26.000000000","submitted":"2023-05-30 18:35:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884718","meta_rev_id":"a3507811ed1ddf94539c36dee762b8b645dc8d47","_number":884718,"virtual_id_number":884718,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 18:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-05-30 16:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-05-30 15:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-30 18:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 18:21:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 15:56:48.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2023-05-30 16:04:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 16:39:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-05-30 18:21:41.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d04203d8d8ceeff65d9f7a8cf4c1ff8c11360e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-05-30 15:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a39a938a4d9445f33c815ff75d9e769b1f92b76","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2023-05-30 15:56:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fc93bbd45b3f0b791c75b5b5953867f195accbd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 16:04:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ba5a0b339f4a8fa6853e5b3545166d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6787a57a60bf46f7b9ececce7c390159 : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8fbeffb2912444039dc05a92fb792cd1 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"aa68fb0305f50bbec9074d878d542f5eff8ea806","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-30 16:39:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2749dbcf102a0d715924baba5aed3129e8c665e2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-30 18:21:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI\u0027mm set the three old repos to inactive and adjust their descriptions once this deploys.","accounts_in_message":[],"_revision_number":1},{"id":"01874fbd828b95b4fc023ff7cd9ae34864946eb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:22:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ef2fb3097f7d1af8455c49044f4aa44c69ddc1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:35:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d48709313d54f299712bcf17d1ddaae\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ff1a36807d4948278ea9fade253504ad : SUCCESS in 7m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4219e8f1fbc7464ab37ca87568b9e689 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"afdb7ea84afda1837eb8ca68b2b529f2d153f180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:35:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"945d99d15ff87ff65fe3a9b145eb3e7f19bfc6c4","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1438022fb6334dc99829b12790fcc2c0\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/def18497314445668d4f17a8e020d293 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a3507811ed1ddf94539c36dee762b8b645dc8d47","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-30 21:09:26.000000000","message":"Patch Set 1:\n\nI\u0027ve updated the descriptions of the three retired repositories in StoryBoard and set them to inactive status.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b3f46d37fe72f1ea7b520b5070fb50b54d419d9","revisions":{"4b3f46d37fe72f1ea7b520b5070fb50b54d419d9":{"kind":"REWORK","_number":1,"created":"2023-05-30 15:41:52.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/18/884718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/884718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/884718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/884718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/884718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/884718/1"}}},"commit":{"parents":[{"commit":"f9b86f0a836187ddd6cc4b223a867abaf1e25ede","subject":"Merge \"Revert \"Pull grafyaml from quay.io\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f9b86f0a836187ddd6cc4b223a867abaf1e25ede"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-05-30 15:31:25.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-05-30 15:41:36.000000000","tz":180},"subject":"Remove retied projects from openstack-helm group","message":"Remove retied projects from openstack-helm group\n\nAt the moment we have 3 active projects in openstack-helm\nproject group. These are\n- openstack-helm\n- openstack-helm-infra\n- openstack-helm-images\nBut we used to have 3 more projects in the project group\nwhich are now retired. These retied projects are\n- openstack-helm-docs\n- openstack-helm-addons\n- openstack-helm-deployments\nNow we would like to cleanup the openstack-helm\nStoryboard project group. We have closed and moved all the\nstories from these retired projects to the main OSH project.\n\nChange-Id: Ib60099f79de902906b8d1956cc202aeb86587cb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b3f46d37fe72f1ea7b520b5070fb50b54d419d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b3f46d37fe72f1ea7b520b5070fb50b54d419d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
