)]}'
{"id":"openstack%2Fproject-config~887941","triplet_id":"openstack%2Fproject-config~master~I095f44c9f934b0532f08dc411700ebd7528cee67","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"35858":{"account":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"last_update":"2023-07-24 12:20:15.000000000","reason":"Change was submitted"},"24777":{"account":{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},"last_update":"2023-07-24 12:20:15.000000000","reason":"Change was submitted"},"28558":{"account":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"last_update":"2023-07-24 12:20:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I095f44c9f934b0532f08dc411700ebd7528cee67","subject":"Add Intel Ethernet Operator app to StarlingX","status":"MERGED","created":"2023-07-07 10:23:22.000000000","updated":"2023-07-24 12:31:31.000000000","submitted":"2023-07-24 12:20:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887941","meta_rev_id":"b6d7dac0ba9f8d2fc061713f842afaab0ad0eecb","_number":887941,"virtual_id_number":887941,"owner":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 12:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-07 12:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-24 12:08:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 10:24:21.000000000","updated_by":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"REVIEWER"},{"updated":"2023-07-07 10:24:21.000000000","updated_by":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"reviewer":{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},"state":"REVIEWER"},{"updated":"2023-07-07 10:37:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 12:30:10.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4b00193ba0595682b761a7942081827831b4bc7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"date":"2023-07-07 10:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25fcb8ff3e50b55fc2f7d4a22908cdab2286ca0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 10:37:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb2db91583440c6862ab2f55f00f07e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6291246272ed4f939aa417a0adce2895 : SUCCESS in 5m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aa8426d28ad64988b1ebee640fbff9b3 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"041090b4b2170e77c6354a916f77d3d13b214a92","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-07 12:30:10.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks fine, though presumably this should wait for formal approval of the associated I659aac562b07c644e6b48f26755b5c664075027b spec change?","accounts_in_message":[],"_revision_number":1},{"id":"61d17bd1d00ecabc98f45fa2005ed248be741afc","author":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"date":"2023-07-07 12:41:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71e27e7da701cef66925e9faf85f8018f1ebee1e","author":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"date":"2023-07-24 09:18:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfa42f81548addd0730eaad25f097372084d0490","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-24 12:08:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cda9ff013463fb3596be6455f26302baaa7aa345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:08:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a0e955dee435d68ba77df43759203c051b936b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:20:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c35de5ec2e842d3a75fe249e831e7ce\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3321aa1806814d7b8567981696be4d20 : SUCCESS in 5m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c6b891c268874c5c83d3dfb236ab78f5 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bb242a36ad39d4b91d719c3c47e6afe81a881a81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:20:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b6d7dac0ba9f8d2fc061713f842afaab0ad0eecb","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:31:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9642ead22041c18c0c94c829a42432\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a2ddf4e762a2472ca62ff4620eca3fb5 : SUCCESS in 5m 06s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/7143d9276c2d4631962a9ba3bf21411a : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d2babc51c4d31b4b29f2f7c0fecf194393122e2","revisions":{"2d2babc51c4d31b4b29f2f7c0fecf194393122e2":{"kind":"REWORK","_number":1,"created":"2023-07-07 10:23:22.000000000","uploader":{"_account_id":35858,"name":"Rafal Lal","display_name":"Rafal Lal","email":"rafalx.lal@intel.com","username":"rafal-lal","status":"Intel"},"ref":"refs/changes/41/887941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/887941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/887941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/887941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/887941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/887941/1"}}},"commit":{"parents":[{"commit":"39d0b03ce6cbc0f4ee188b339322bd0b893e028e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39d0b03ce6cbc0f4ee188b339322bd0b893e028e"}]}],"author":{"name":"Rafal Lal","email":"rafalx.lal@intel.com","date":"2023-07-07 10:21:17.000000000","tz":120},"committer":{"name":"Rafal Lal","email":"rafalx.lal@intel.com","date":"2023-07-07 10:21:17.000000000","tz":120},"subject":"Add Intel Ethernet Operator app to StarlingX","message":"Add Intel Ethernet Operator app to StarlingX\n\nStory: 2010562\nTask: 48334\n\nChange-Id: I095f44c9f934b0532f08dc411700ebd7528cee67\nSigned-off-by: Rafal Lal \u003crafalx.lal@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d2babc51c4d31b4b29f2f7c0fecf194393122e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2d2babc51c4d31b4b29f2f7c0fecf194393122e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
