)]}'
{"id":"openstack%2Fproject-config~888066","triplet_id":"openstack%2Fproject-config~master~Ia6fa05f6aead94a0e70376e2c51d9b551a189660","project":"openstack/project-config","branch":"master","topic":"acl-charms.ceph","attention_set":{},"removed_from_attention_set":{"8992":{"account":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"last_update":"2023-07-25 20:48:10.000000000","reason":"Change was submitted"},"33717":{"account":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"last_update":"2023-07-25 20:48:10.000000000","reason":"Change was submitted"},"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2023-07-14 15:58:12.000000000","reason":"\u003cGERRIT_ACCOUNT_2424\u003e replied on the change","reason_account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-07-14 15:52:50.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ia6fa05f6aead94a0e70376e2c51d9b551a189660","subject":"Add ceph-related projects to charm-ceph acl","status":"MERGED","created":"2023-07-10 19:19:15.000000000","updated":"2023-07-25 20:53:48.000000000","submitted":"2023-07-25 20:48:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888066-acl-charms.ceph","meta_rev_id":"bab45d7f2609e9b5f3636c430780c973e5bc632d","_number":888066,"virtual_id_number":888066,"owner":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-25 20:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2023-07-14 15:58:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-25 20:32:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-07-14 15:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2023-07-25 20:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"}],"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-10 19:26:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-07-10 19:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-14 15:36:37.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"CC"},{"updated":"2023-07-14 15:36:37.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"state":"CC"},{"updated":"2023-07-14 15:36:46.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"},{"updated":"2023-07-25 20:32:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"a155c0016795b0fb6fbd28a7b65e64cdde6093f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2023-07-10 19:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ce15bfd201defc6299fe244f41774f0196b8214","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-10 19:26:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nIt would be good to get Felipe (OpenStack Charms PTL) to +1 this before we approve it.","accounts_in_message":[],"_revision_number":1},{"id":"2550340845f012a3f54e59cdb4cdd368284db02d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 19:27:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4ade1b730a4f9d8daabc4c1eb2c535\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b2c2a59bd9f74bf38851f74c4485658d : SUCCESS in 6m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/36042233dc364f0dbc76bba5fe47282a : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dc9ad16509e0f6f6afc223b5cddf1b64dfd597ae","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-07-14 15:36:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cbafc5a4b844fff75116b494617c228a5aada48","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-07-14 15:36:46.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"6750c569b18f23cfb2a8cc67f31fdb6f1136f2f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2023-07-14 15:40:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c280b0e754643dd65adc43165ca09aef97a62eed","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-07-14 15:42:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"938b51b97e8656c4d483982d89d5d30898e8a436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 15:49:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4928a86bf186403186d3e99a5f143583\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/264b16412af04d33bae31805a0e59086 : SUCCESS in 8m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1927fa1075044b2eb81e4d17a72259fa : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"81abc6e8426d4193f013fe0cf46e2ad3b7cda144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2023-07-14 15:51:33.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"256c37e9c3d3a132c47d76d0be080f68a7d26ece","author":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"date":"2023-07-14 15:51:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"59298757a77e8e4cff71d222e0a4eed1618c32cb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-14 15:52:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d74d3b67085f452a3265c6d7cf6002ea8581b43a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-14 15:53:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3374da19881e3e69edf20103bd5e4ac9f9832f46","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-14 15:55:18.000000000","message":"Patch Set 3:\n\nLooks like my comment about the wrong commit ID mentioned in the commit message was also something I spotted on the first version of the patch, I guess it got overlooked.","accounts_in_message":[],"_revision_number":3},{"id":"51d35f1517d6d1231ebb144ba0a5b3acda21733b","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-07-14 15:58:12.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22ba5779d62e428a8d0b88555aaa7cacbcd9487e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 15:59:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db8539f7a8b14f638ef658a95d3c20ba\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aef0df5c503a4c18ad117d40f42d5b79 : SUCCESS in 6m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d9021db762142879c2eb9b86433bcdc : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"92b0cceb284a4bfa2f9c1630167c64982bfd2f31","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-25 20:32:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a448063d8bc368b03109684c962ac623722a27f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-25 20:36:20.000000000","message":"Patch Set 3: Workflow+1\n\nTreating the incorrect commit ID reference in the commit message as a nit, and approving to get this moving along.","accounts_in_message":[],"_revision_number":3},{"id":"6528a19f8c0c7b5b9184bd199d9b0b8e872c5c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:36:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c47a63caf259bc470734191ae9ccca1dc9afd4b2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-25 20:41:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d37353c278dcf08ec98d8706d82bb16af8505956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:48:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567e1289efe542cf9c2e7ed51e4e4c46\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/018a4d1dbc374e62b15c501fab90988f : SUCCESS in 6m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/266bcb75f6d24dcfb711f649727cce82 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1a768540fc8eb1dc8d199e14383f88c40a033f68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:48:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bab45d7f2609e9b5f3636c430780c973e5bc632d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:53:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a648b7838e49cdab91480e5fb04ad5\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f5ee6bb09ac84991807aa4b51fd58f1a : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"69b27e0dee60557be76ce7b0dcd6dd6dbee0d8c5","revisions":{"88a0ed5529aa90b9baf998818019cd7ffc1183a2":{"kind":"REWORK","_number":1,"created":"2023-07-10 19:19:15.000000000","uploader":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"ref":"refs/changes/66/888066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/888066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/888066/1"}}},"commit":{"parents":[{"commit":"39d0b03ce6cbc0f4ee188b339322bd0b893e028e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39d0b03ce6cbc0f4ee188b339322bd0b893e028e"}]}],"author":{"name":"Billy Olsen","email":"billy.olsen@gmail.com","date":"2023-07-10 19:17:31.000000000","tz":-420},"committer":{"name":"Billy Olsen","email":"billy.olsen@gmail.com","date":"2023-07-10 19:19:13.000000000","tz":-420},"subject":"Add charms.ceph to charm-ceph acl","message":"Add charms.ceph to charm-ceph acl\n\nThe ceph charms were broken out into their own project ACL with\ncommit 64afff5b73 but it missed the charms.ceph project which is\na common library for ceph charms. This patch fixes the ACL to point\nto the charm-ceph.config for the acl list for charms.ceph.\n\nChange-Id: Ia6fa05f6aead94a0e70376e2c51d9b551a189660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a0ed5529aa90b9baf998818019cd7ffc1183a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a0ed5529aa90b9baf998818019cd7ffc1183a2"}]},"branch":"refs/heads/master"},"3ce765604e9df093f23032fc0202e24c876b7919":{"kind":"REWORK","_number":2,"created":"2023-07-14 15:40:10.000000000","uploader":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"ref":"refs/changes/66/888066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/888066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/888066/2"}}},"commit":{"parents":[{"commit":"39d0b03ce6cbc0f4ee188b339322bd0b893e028e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39d0b03ce6cbc0f4ee188b339322bd0b893e028e"}]}],"author":{"name":"Billy Olsen","email":"billy.olsen@gmail.com","date":"2023-07-10 19:17:31.000000000","tz":-420},"committer":{"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","date":"2023-07-14 15:39:16.000000000","tz":-180},"subject":"Add charms.ceph to charm-ceph acl","message":"Add charms.ceph to charm-ceph acl\n\nThe ceph charms were broken out into their own project ACL with\ncommit 64afff5b73 but it missed a few ceph-related projects.\nThis patch fixes the ACL to point to the charm-ceph.config for the\nacl list for all these projects.\n\nChange-Id: Ia6fa05f6aead94a0e70376e2c51d9b551a189660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ce765604e9df093f23032fc0202e24c876b7919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ce765604e9df093f23032fc0202e24c876b7919"}]},"branch":"refs/heads/master"},"69b27e0dee60557be76ce7b0dcd6dd6dbee0d8c5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-07-14 15:51:33.000000000","uploader":{"_account_id":33717,"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","username":"lmlg"},"ref":"refs/changes/66/888066/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/888066/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/888066/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/888066/3"}}},"commit":{"parents":[{"commit":"39d0b03ce6cbc0f4ee188b339322bd0b893e028e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39d0b03ce6cbc0f4ee188b339322bd0b893e028e"}]}],"author":{"name":"Billy Olsen","email":"billy.olsen@gmail.com","date":"2023-07-10 19:17:31.000000000","tz":-420},"committer":{"name":"Luciano Lo Giudice","email":"luciano.logiudice@canonical.com","date":"2023-07-14 15:51:14.000000000","tz":-180},"subject":"Add ceph-related projects to charm-ceph acl","message":"Add ceph-related projects to charm-ceph acl\n\nThe ceph charms were broken out into their own project ACL with\ncommit 64afff5b73 but it missed a few ceph-related projects.\nThis patch fixes the ACL to point to the charm-ceph.config for the\nacl list for all these projects.\n\nChange-Id: Ia6fa05f6aead94a0e70376e2c51d9b551a189660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69b27e0dee60557be76ce7b0dcd6dd6dbee0d8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/69b27e0dee60557be76ce7b0dcd6dd6dbee0d8c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
