)]}'
{"id":"openstack%2Fproject-config~888467","triplet_id":"openstack%2Fproject-config~master~Idf69449c3e107bbaf4f853bce0cb160464cbe59a","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2023-07-19 12:11:42.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-07-19 11:57:18.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-07-19 12:11:42.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-07-13 16:12:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-07-13 14:52:00.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Idf69449c3e107bbaf4f853bce0cb160464cbe59a","subject":"Allow QA cores to toggle WIP state","status":"MERGED","created":"2023-07-13 13:46:47.000000000","updated":"2023-07-19 12:25:38.000000000","submitted":"2023-07-19 12:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888467","meta_rev_id":"c5abc72ea1c834fd80654c252ba51c7670ff4db0","_number":888467,"virtual_id_number":888467,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-19 12:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-07-19 11:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-13 16:12:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-07-13 14:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-07-19 11:57:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 13:47:31.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-07-13 13:47:31.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-07-13 13:47:31.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-07-13 13:59:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-17 06:39:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"}],"messages":[{"id":"6827a33cc77f4b69e372cb6d35be58c7542ecdd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-13 13:46:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35a17391a95f8dbdff0b90c468acc698447c7f89","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-13 13:54:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e818684da1f4cd393fee9200aa42b448e4703be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 13:59:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0f592ce730945f2ad886a8e9c75b8d9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3fc7c20892444a3e995000e3b502e66d : SUCCESS in 6m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/00d71bd1cfc344d189dbdb7b341264c4 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5ebfc4e06a692075d72a2c6c89ee7d1d05ebcbcd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-13 13:59:44.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nMinor concern marked inline. Also for completeness we should probably get the okay from QA leadership (PTL or Infra Liaison) before approving.","accounts_in_message":[],"_revision_number":1},{"id":"06ef61948941ef8275432ae7ba058c9ba595b725","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-13 14:36:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1e47c077fe04c00713945b7f2b4716abe673ea3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-13 14:37:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0036a6f203ff5cfaffdaa78fc86d5bf03a0dcfed","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-13 14:43:56.000000000","message":"Patch Set 2: Code-Review-1\n\nYou need to include the \"group\" keyword. See Ibae9274127e904337f57813974f5745c41980c9e for details.","accounts_in_message":[],"_revision_number":2},{"id":"71313dc933cf6fde014066a79c022b014c99c3c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 14:45:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4b7b51447c4dc0b96a1b7546691cdf\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8a1ffcc0a5964369a949ed8c4484f345 : SUCCESS in 6m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5b281d7273734558ab99f0fffde4cf2a : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a5e01e7d966bbc77b6695de5e43dd5855c91ed92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-13 14:50:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"899f8d91661c90607cdb477ba5cc703d3efbdcaa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-13 14:52:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"50db77103955110cd3fafe3c89fae462e619523f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 15:10:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6296e6700a4b2c8307aea4ff545a91\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca0b2d55f72d4fa09d7a34537027f037 : SUCCESS in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c7ec3ed78a9f4c86b29f0d68cc8437b9 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"bc37bc04808dd930fc7c612f812158efa4e571d6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-13 16:12:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d477222e390c9ab4f15c27c251053d665eb471d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-07-17 06:39:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebf15f49c451cb83a03f52084147d6bb5d5b947f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-07-17 09:16:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"041f4a4ae934ecd55e1a41b3e2bb51bbe76300d2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-07-19 11:57:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55a4a9255016a201c16f9d0b4ff0bfdbac554317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 11:57:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86d827e9d9e191a808513852a61735bdc102c4fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 12:11:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/172687eee67f4f17ae48c0583d263226\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/01ba44c25a724f00ba74141e30e214d2 : SUCCESS in 6m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ad26822257bb48daab36c447e01d276d : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"237530be6a98d3027863c34e6c86326fc0ac1946","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 12:11:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c5abc72ea1c834fd80654c252ba51c7670ff4db0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 12:25:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/651bad0282b046ad93be7b71ed99c011\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/47e1c5aa255e49cba1810c92c4c6d0f0 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2da44a6c815eaeac23c2220011cb6c6deb8ee815","revisions":{"32968a4a4bc41a00a3af17fdeb1df3885c6c249d":{"kind":"REWORK","_number":1,"created":"2023-07-13 13:46:47.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/67/888467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/888467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/888467/1"}}},"commit":{"parents":[{"commit":"27161af24ad42c71af62f432dd6868d7a5be3803","subject":"Allow refstack cores to toggle WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27161af24ad42c71af62f432dd6868d7a5be3803"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-07-13 13:45:52.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-07-13 13:45:52.000000000","tz":120},"subject":"Allow QA cores to toggle WIP state","message":"Allow QA cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: Idf69449c3e107bbaf4f853bce0cb160464cbe59a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32968a4a4bc41a00a3af17fdeb1df3885c6c249d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32968a4a4bc41a00a3af17fdeb1df3885c6c249d"}]},"branch":"refs/heads/master"},"dd4a75b1aab45013d1008c808d7f139ef4adaa31":{"kind":"REWORK","_number":2,"created":"2023-07-13 14:37:22.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/67/888467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/888467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/888467/2"}}},"commit":{"parents":[{"commit":"27161af24ad42c71af62f432dd6868d7a5be3803","subject":"Allow refstack cores to toggle WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27161af24ad42c71af62f432dd6868d7a5be3803"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-07-13 13:45:52.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-07-13 14:35:57.000000000","tz":120},"subject":"Allow QA cores to toggle WIP state","message":"Allow QA cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: Idf69449c3e107bbaf4f853bce0cb160464cbe59a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd4a75b1aab45013d1008c808d7f139ef4adaa31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd4a75b1aab45013d1008c808d7f139ef4adaa31"}]},"branch":"refs/heads/master"},"2da44a6c815eaeac23c2220011cb6c6deb8ee815":{"kind":"REWORK","_number":3,"created":"2023-07-13 14:50:51.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/67/888467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/67/888467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/67/888467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/67/888467/3"}}},"commit":{"parents":[{"commit":"27161af24ad42c71af62f432dd6868d7a5be3803","subject":"Allow refstack cores to toggle WIP state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/27161af24ad42c71af62f432dd6868d7a5be3803"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-07-13 13:45:52.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-07-13 14:50:45.000000000","tz":120},"subject":"Allow QA cores to toggle WIP state","message":"Allow QA cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: Idf69449c3e107bbaf4f853bce0cb160464cbe59a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2da44a6c815eaeac23c2220011cb6c6deb8ee815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2da44a6c815eaeac23c2220011cb6c6deb8ee815"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
