)]}'
{"id":"openstack%2Fproject-config~888901","triplet_id":"openstack%2Fproject-config~master~Icfd58a88a12d13f093c08f41ab4be85c26051149","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2023-07-25 20:05:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfd58a88a12d13f093c08f41ab4be85c26051149","subject":"Add 32GB nodes Ubuntu Focal and Jammy nodes","status":"MERGED","created":"2023-07-18 23:15:03.000000000","updated":"2023-07-25 20:28:19.000000000","submitted":"2023-07-25 20:05:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888901","meta_rev_id":"2cd339710f0d2a10390e9c9822be83b72894bc83","_number":888901,"virtual_id_number":888901,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-25 20:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-07-25 19:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-19 17:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-07-25 19:53:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-18 23:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 00:10:08.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-07-19 17:16:52.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-07-25 19:53:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"b70d22fb30752d8ef7ad064af0df38d5d8d1dd5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-18 23:15:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08cf937a8c2f3e1f9eb49d99982f3d18796aaf3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-18 23:17:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b0276b28cbad680b8db72d180bce8cbf431cbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-18 23:19:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f17a80914ef9ab7003a808f36754f0e465dbb3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-18 23:21:08.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"de5f12718b39709852863f69606d900aef3b8ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 23:30:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13abaafba79d4283878aaf68f9547e49\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/10bce45615684f71a857c08f3607548d : SUCCESS in 6m 32s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/594f170a4d344e0cb618d750d59918b7 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"e4dedcbd748922e0bb8b0d786799834f864e7a13","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-19 00:10:08.000000000","message":"Patch Set 4:\n\nFor efficiency of packing nodes into our available quota at peak load, we prefer to stick to standard sizes for frequently run jobs. Consider redesigning the job to take advantage of a multi-node nodeset and distributing the work across more than one test node.\n\nWhile it\u0027s not clear from the configuration, the v3-standard-8 flavor in vexxhost ca-ymq-1 has 32GB RAM (this is why we don\u0027t use that region to supply any of our standard labels). You could start out creating variations of the existing ubuntu-bionic-vexxhost label for other images just in that provider/region on a trial basis with limited impact to efficiency in our other environments.","accounts_in_message":[],"_revision_number":4},{"id":"e8d26b4d71c21372947c4495743ffced19a06b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-19 11:53:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9aca21e4713dad3edda7c0b2d70699b942f061dd","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-19 11:57:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11cfeaab711bdfb7775ba0af02029e0ca61044bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 12:06:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d836beb21964a38b4f06b9e5d55b509\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f698e295e2d245a788c989d818b38f41 : SUCCESS in 6m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/263542947ae147ed9730bb2b6794e11c : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":5},{"id":"0159be3ccd9452b52c1123d64eae9e4fa3abf496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2023-07-19 12:42:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3321e51ae76c9d62367c9ce55e0e10fd55eda14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 12:58:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b58070e53c443ab17b4f76d2e823fc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4310bc7d5d5496681e93640f2e0a59e : SUCCESS in 6m 09s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1774134ea1d0479e8edfa3ad3f5c93d8 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":6},{"id":"132e5bbbce0bfc18486204df502ca638f50fb29d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-19 17:16:52.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"fc7fcf063d12a4a12d321ecd6606803e5a4b2cb4","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-07-25 19:53:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3f19062ccaf2a4787668f3505053031f92c5cc63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 19:54:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"47054e1c0e978f20a04a5701089055ab08b4ae3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:05:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331a3fc1859a4f2393d060129a88689f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a290caf7da734ffb80fd7c3cd7e17d3d : SUCCESS in 6m 41s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/cafec767742344e59ba786207ed6cdb3 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":6},{"id":"d33e20e9fbce6d6f57c7b4e6901f81611afdbde7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:05:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"2cd339710f0d2a10390e9c9822be83b72894bc83","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 20:28:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66643c7fb9704fa58c262c1f406696a3\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/c8b5a140a2b9488394f62c66b1adc502 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"50c4046096268f618008f668f4a08755eafa1165","revisions":{"4f85731f3ec39994ab2a7dbe771ee894c486995c":{"kind":"REWORK","_number":1,"created":"2023-07-18 23:15:03.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/01/888901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/888901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/888901/1"}}},"commit":{"parents":[{"commit":"ebd6f60bb733652636e1c026341b42c1b32e6a9a","subject":"Catch missing \"group\" keywords in Gerrit ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebd6f60bb733652636e1c026341b42c1b32e6a9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"subject":"Add 16GB nodes to nodepool providers","message":"Add 16GB nodes to nodepool providers\n\nSome deployment projects (e.g. Openstack-Helm)\ntest their code as \"all-in-one\" deployment.\nI.e. a test job deploys all Openstack components\non a single node sometimes on top of a minimalistic\nK8s cluster running on the same node.\n\nThis sometimes requires more memory than 8GB to make\njobs reliable and we need labels explicitly specifing\nthat a node has at least 16GB memory.\n\nChange-Id: Icfd58a88a12d13f093c08f41ab4be85c26051149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f85731f3ec39994ab2a7dbe771ee894c486995c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4f85731f3ec39994ab2a7dbe771ee894c486995c"}]},"branch":"refs/heads/master"},"bc30a0450d32175f43e90ee1096ee4583d3b562f":{"kind":"REWORK","_number":2,"created":"2023-07-18 23:17:41.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/01/888901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/888901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/888901/2"}}},"commit":{"parents":[{"commit":"ebd6f60bb733652636e1c026341b42c1b32e6a9a","subject":"Catch missing \"group\" keywords in Gerrit ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebd6f60bb733652636e1c026341b42c1b32e6a9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:17:31.000000000","tz":180},"subject":"Add 16GB nodes to nodepool providers","message":"Add 16GB nodes to nodepool providers\n\nSome deployment projects (e.g. Openstack-Helm)\ntest their code as \"all-in-one\" deployment.\nI.e. a test job deploys all Openstack components\non a single node sometimes on top of a minimalistic\nK8s cluster running on the same node.\n\nThis sometimes requires more memory than 8GB to make\njobs reliable and we need labels explicitly specifing\nthat a node has at least 16GB memory.\n\nChange-Id: Icfd58a88a12d13f093c08f41ab4be85c26051149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc30a0450d32175f43e90ee1096ee4583d3b562f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bc30a0450d32175f43e90ee1096ee4583d3b562f"}]},"branch":"refs/heads/master"},"1e1437444cebd66ace17e8bf25c40c80a91d33e0":{"kind":"REWORK","_number":3,"created":"2023-07-18 23:19:29.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/01/888901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/888901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/888901/3"}}},"commit":{"parents":[{"commit":"ebd6f60bb733652636e1c026341b42c1b32e6a9a","subject":"Catch missing \"group\" keywords in Gerrit ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebd6f60bb733652636e1c026341b42c1b32e6a9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:19:20.000000000","tz":180},"subject":"Add 16GB nodes to nodepool providers","message":"Add 16GB nodes to nodepool providers\n\nSome deployment projects (e.g. Openstack-Helm)\ntest their code as \"all-in-one\" deployment.\nI.e. a test job deploys all Openstack components\non a single node sometimes on top of a minimalistic\nK8s cluster running on the same node.\n\nThis sometimes requires more memory than 8GB to make\njobs reliable and we need labels explicitly specifing\nthat a node has at least 16GB memory.\n\nChange-Id: Icfd58a88a12d13f093c08f41ab4be85c26051149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e1437444cebd66ace17e8bf25c40c80a91d33e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1e1437444cebd66ace17e8bf25c40c80a91d33e0"}]},"branch":"refs/heads/master"},"6d0875cf8e0c0715226ba1a2fb51e79e78eb70ca":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-07-18 23:21:08.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/01/888901/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/888901/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/888901/4"}}},"commit":{"parents":[{"commit":"ebd6f60bb733652636e1c026341b42c1b32e6a9a","subject":"Catch missing \"group\" keywords in Gerrit ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebd6f60bb733652636e1c026341b42c1b32e6a9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:20:41.000000000","tz":180},"subject":"Add 16GB nodes to nodepool providers","message":"Add 16GB nodes to nodepool providers\n\nSome deployment projects (e.g. Openstack-Helm)\ntest their code as \"all-in-one\" deployment.\nI.e. a test job deploys all Openstack components\non a single node on top of a minimalistic\nK8s cluster running on the same node.\n\nThis requires more memory than 8GB to make\njobs reliable and we need labels explicitly specifing\nthat a node has at least 16GB memory.\n\nChange-Id: Icfd58a88a12d13f093c08f41ab4be85c26051149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d0875cf8e0c0715226ba1a2fb51e79e78eb70ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d0875cf8e0c0715226ba1a2fb51e79e78eb70ca"}]},"branch":"refs/heads/master"},"02a232a0bf26056e11cc1fc16053ec9c36eabc74":{"kind":"REWORK","_number":5,"created":"2023-07-19 11:53:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/01/888901/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/888901/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/888901/5"}}},"commit":{"parents":[{"commit":"ebd6f60bb733652636e1c026341b42c1b32e6a9a","subject":"Catch missing \"group\" keywords in Gerrit ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebd6f60bb733652636e1c026341b42c1b32e6a9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-19 11:44:51.000000000","tz":180},"subject":"Add 32GB nodes Ubuntu Focal and Jammy nodes","message":"Add 32GB nodes Ubuntu Focal and Jammy nodes\n\nSome deployment projects (e.g. Openstack-Helm)\ntest their code as \"all-in-one\" deployment.\nI.e. a test job deploys all Openstack components\non a single node on top of a minimalistic\nK8s cluster running on the same node.\n\nThis requires more memory than 8GB to make\njobs reliable. We add these new *-32GB labels\nonly in the Vexxhost ca-ymq-1 region because\nv3-standard-8 flavor in this region has 32GB\nnodes.\n\nAt the same time we should not stick to this\nkind of nodes since the number of such nodes\nis very limited. It is highly recommended to\nredesign the test jobs so they use multinode\nnodesets.\n\nChange-Id: Icfd58a88a12d13f093c08f41ab4be85c26051149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/02a232a0bf26056e11cc1fc16053ec9c36eabc74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/02a232a0bf26056e11cc1fc16053ec9c36eabc74"}]},"branch":"refs/heads/master"},"50c4046096268f618008f668f4a08755eafa1165":{"kind":"REWORK","_number":6,"created":"2023-07-19 12:42:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/01/888901/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/888901/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/888901/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/888901/6"}}},"commit":{"parents":[{"commit":"ebd6f60bb733652636e1c026341b42c1b32e6a9a","subject":"Catch missing \"group\" keywords in Gerrit ACL files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebd6f60bb733652636e1c026341b42c1b32e6a9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-18 23:06:23.000000000","tz":180},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2023-07-19 12:42:11.000000000","tz":180},"subject":"Add 32GB nodes Ubuntu Focal and Jammy nodes","message":"Add 32GB nodes Ubuntu Focal and Jammy nodes\n\nSome deployment projects (e.g. Openstack-Helm)\ntest their code as \"all-in-one\" deployment.\nI.e. a test job deploys all Openstack components\non a single node on top of a minimalistic\nK8s cluster running on the same node.\n\nThis requires more memory than 8GB to make\njobs reliable. We add these new *-32GB labels\nonly in the Vexxhost ca-ymq-1 region because\nv3-standard-8 flavor in this region has 32GB\nnodes.\n\nAt the same time we should not stick to this\nkind of nodes since the number of such nodes\nis very limited. It is highly recommended to\nredesign the test jobs so they use multinode\nnodesets.\n\nChange-Id: Icfd58a88a12d13f093c08f41ab4be85c26051149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50c4046096268f618008f668f4a08755eafa1165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50c4046096268f618008f668f4a08755eafa1165"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
