)]}'
{"id":"openstack%2Fproject-config~888930","triplet_id":"openstack%2Fproject-config~master~I13d8434d78ef7839eb66149fe4b6f8454b238ccb","project":"openstack/project-config","branch":"master","topic":"apt-ostree","attention_set":{},"removed_from_attention_set":{"24":{"account":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"last_update":"2023-07-21 16:21:28.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-07-21 16:10:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I13d8434d78ef7839eb66149fe4b6f8454b238ccb","subject":"Add starlingx/apt-ostree","status":"MERGED","created":"2023-07-19 14:56:27.000000000","updated":"2023-07-25 14:27:04.000000000","submitted":"2023-07-21 16:21:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888930-apt-ostree","meta_rev_id":"e4aa0641d0726e121f567644b009343d605e431f","_number":888930,"virtual_id_number":888930,"owner":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-21 16:21:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},"all":[{"value":1,"date":"2023-07-25 14:27:04.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-19 17:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-21 16:10:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 15:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 17:15:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-07-25 14:27:04.000000000","updated_by":{"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},"reviewer":{"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},"state":"REVIEWER"}],"messages":[{"id":"d4972f3be2bb4d834176ee1aaba9e863d42a0e6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2023-07-19 14:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8b1c060e14ab55fffd498b07a41340b8c7ffa13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:04:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0606df2ec8484ed9b8188340681ec1d3\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b2ca02a255f84cbbb77aad85cd1fd58e : SUCCESS in 6m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/eb2bb89221234c469bf6dc14d26e5352 : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2bb2cdf32ab7bb03f476d8a7449287197987322d","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2023-07-19 15:36:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43ad01e3354246d70844a9e114b3b8ad25fe73c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:53:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a584465a99674562bf4723aa1d607f4e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd8cddbe17774dc29da898ebe714d560 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7e5117b1a5a4430e946fd6e7efef7fdc : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"514b215a00087e42b3695188d15aeb8cafd028b8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-19 17:15:51.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks technically correct. Discussion of this proposal on the StarlingX mailing list seems to be in progress, which is presumably the reason for the WIP flag.","accounts_in_message":[],"_revision_number":1},{"id":"90e25b9cadc1d3e36c7b0fe2d56db6fba4992700","author":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"date":"2023-07-20 14:04:02.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"d7a70b5d0cea71e4287cc0cf265f9be938af75b5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-07-21 16:10:13.000000000","message":"Patch Set 1: Workflow+1\n\nSingle-core approving this as there seems (based on limited feedback on the ML) to be consensus from project leadership, and the usual infra liaison for StarlingX is still on holiday.","accounts_in_message":[],"_revision_number":1},{"id":"f7650f19049bfb43161d13146703950ca88765d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 16:10:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"242ae97e8f7f907c32bb8b8ef33284509692d553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 16:21:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b847c499a12141008ba06c559eb26ae1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f6175838b6be47468a8aa86a496d569c : SUCCESS in 6m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8a91168adcbb47b9b9a0fe679998e5c1 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3425fb36b896f0990893b83657b81743bec53dcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 16:21:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be9ec1386373853fa383dbcc073b5f28c1418969","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 16:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91a6ea57a8d64279bc9aba87c5a348f8\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/e3ed048523e94184a916d6edc035c08f : SUCCESS in 5m 13s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ece9f32d0af349bd9b3e7ec1aa26a63d : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e4aa0641d0726e121f567644b009343d605e431f","author":{"_account_id":9926,"name":"John Kung","email":"john.kung@windriver.com","username":"jkung"},"date":"2023-07-25 14:27:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37a7137ec17c1549eda5c84a2487b52efee0c267","revisions":{"37a7137ec17c1549eda5c84a2487b52efee0c267":{"kind":"REWORK","_number":1,"created":"2023-07-19 14:56:27.000000000","uploader":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"ref":"refs/changes/30/888930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/888930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/888930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/888930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/888930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/888930/1"}}},"commit":{"parents":[{"commit":"c5540202ba6bea926e92ad0dc52509cbdac7a73a","subject":"Merge \"Allow QA cores to toggle WIP state\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5540202ba6bea926e92ad0dc52509cbdac7a73a"}]}],"author":{"name":"Charles Short","email":"charles.short@windriver.com","date":"2023-07-19 14:51:39.000000000","tz":-240},"committer":{"name":"Charles Short","email":"charles.short@windriver.com","date":"2023-07-19 14:56:26.000000000","tz":-240},"subject":"Add starlingx/apt-ostree","message":"Add starlingx/apt-ostree\n\nInitial repository code extracted from\nhttps://github.com/zulcss/apt-ostree.\n\nTask: 2010836\n\nChange-Id: I13d8434d78ef7839eb66149fe4b6f8454b238ccb\nSigned-off-by: Charles Short \u003ccharles.short@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37a7137ec17c1549eda5c84a2487b52efee0c267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37a7137ec17c1549eda5c84a2487b52efee0c267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
