)]}'
{"id":"openstack%2Fproject-config~890312","triplet_id":"openstack%2Fproject-config~master~I7721ea6471577a6636e4df7db656eb3cc39eed8c","project":"openstack/project-config","branch":"master","topic":"sunbeam-groups","attention_set":{},"removed_from_attention_set":{"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2023-08-02 11:45:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7721ea6471577a6636e4df7db656eb3cc39eed8c","subject":"Sunbeam: split ACL\u0027s from OpenStack Charms","status":"MERGED","created":"2023-08-02 08:54:55.000000000","updated":"2023-08-02 11:50:58.000000000","submitted":"2023-08-02 11:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890312-sunbeam-groups","meta_rev_id":"bc43df8342128e352ed58ccf17effce8f5ca2542","_number":890312,"virtual_id_number":890312,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 11:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-02 11:37:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-02 11:37:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 09:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 11:37:12.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"eb29fad0cc9915a97276fef6deaa93d91cbb9f16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-08-02 08:54:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"425d2582cea976a8fdba19970471c830882fa4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 09:02:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79223ef3e04641e7ab21f6bf160b5d44\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9ff710465304ce8af70d781a3c55892 : SUCCESS in 6m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/68d28a2043fb4de7b8bea23fb3bd1723 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9e84569112f631c82018c20eaab317a8a9077014","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-02 11:37:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis matches what\u0027s in governance (aside from the two repos in the x namespace of course). Once this deploys I\u0027ll add you as the initial member of the new groups in Gerrit.","accounts_in_message":[],"_revision_number":1},{"id":"16ca103ea8581267706b97f46c3759f72c7b04bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:37:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76eaf5a7d6b00a2a3576e64b63f4d99819af88d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:45:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c2f1e87ea084fad9deda6ab35f7a64a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e0b669ead1c24ffcb5d3d29026d09062 : SUCCESS in 6m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/47e05c12e74e48c7a2cc2807cf3095d2 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e37fffc893bf53f2c92908434641cd204aa7ea39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:45:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bc43df8342128e352ed58ccf17effce8f5ca2542","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19156126dc974727bb9a2a2cf1d0944a\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/3539e7be9bed42b2bb72a47bc86e77fd : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16cd936fdfd15db06c6f03bd84988c4d22b1f7af","revisions":{"16cd936fdfd15db06c6f03bd84988c4d22b1f7af":{"kind":"REWORK","_number":1,"created":"2023-08-02 08:54:55.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/12/890312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/890312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/890312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/890312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/890312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/890312/1"}}},"commit":{"parents":[{"commit":"6e27f80e8c522c34ed69108fb823db73d830e984","subject":"Merge \"Switch from deprecated require-approval to require\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e27f80e8c522c34ed69108fb823db73d830e984"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-08-02 08:51:51.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2023-08-02 08:54:53.000000000","tz":60},"subject":"Sunbeam: split ACL\u0027s from OpenStack Charms","message":"Sunbeam: split ACL\u0027s from OpenStack Charms\n\nAs Sunbeam is now a distinct project, split the ACL\u0027s for all\nproject related repositories from the OpenStack Charms project.\n\nChange-Id: I7721ea6471577a6636e4df7db656eb3cc39eed8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16cd936fdfd15db06c6f03bd84988c4d22b1f7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16cd936fdfd15db06c6f03bd84988c4d22b1f7af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
