)]}'
{"id":"openstack%2Fproject-config~890365","triplet_id":"openstack%2Fproject-config~master~Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-08 12:15:17.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-08-08 12:15:17.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-08-08 08:51:57.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4","subject":"Allow designate-core as osc/sdk service-core","status":"MERGED","created":"2023-08-02 21:04:12.000000000","updated":"2023-08-08 12:31:37.000000000","submitted":"2023-08-08 12:15:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890365","meta_rev_id":"50e4e9ed6fb8ce2a61a724e7383dfe3ab95f4a15","_number":890365,"virtual_id_number":890365,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-08 12:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-08 08:51:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":2,"date":"2023-08-08 12:02:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":1,"date":"2023-08-08 12:31:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2023-08-08 12:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 21:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-03 05:24:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-08-04 11:45:22.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-08-07 22:43:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-08 08:19:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-08-08 12:02:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-08-08 12:31:37.000000000","updated_by":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"reviewer":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"state":"REVIEWER"}],"messages":[{"id":"32fc27d274c0886902025c86c5fcf4bcfe882215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-02 21:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72ddfe1b417c679667676aad09c96355a5b1da21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 21:10:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef497d9832ec48a984b5bb3fbf9a84e2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6ebd00ddda054e6d81ea8a38ca70ea6a : SUCCESS in 5m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fcb9245761d44719a0301140ede45832 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"308dd34ccfa6172858210f04ea80bb24b7d82411","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-03 05:24:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af4a54a46e798b395ee31c9e887bfa61192d0bab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-03 14:43:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b46f4f57a23704229a05eef5af1308c1811e2c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 14:55:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/973d2fd38bf744968114a10aba88468e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/153333ae72914569a107f0659c6dea65 : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/24dd523a0ed04a6ea766d9e12e4b763a : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"06d3f053e3b59bbdd51b2cf85a4c589a6e5089a8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-04 11:45:22.000000000","message":"Patch Set 2:\n\nThe reason for the job failure is that the entries need to be kept in alphabetical order (which will become increasingly useful as this list grows considerably longer).","accounts_in_message":[],"_revision_number":2},{"id":"1309033dc27ffe0fd4adfd5d807856a89804a927","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-07 22:42:43.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1acbcf543f302c54f09d2574068f0c9a8fbf1761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 22:42:59.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/project-config for 890365,3","accounts_in_message":[],"_revision_number":3},{"id":"b2465a5995bcbf1e2deb0ce6dac274b1d53b113d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-07 22:43:02.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/project-config for 890365,3","accounts_in_message":[],"_revision_number":3},{"id":"6b0a705b92e5e57745ac13d70a9bd01f69d9fd9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-07 22:48:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3e1cd5f8566fd948c42a421e5bffc4238f7b0664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 23:03:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ab810cf6f554cbb9c4623db60d6b685\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9291ae44a91a4d62a942726b83871559 : SUCCESS in 6m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3f35e2e11d774ae1bf9b57bab9a27357 : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":4},{"id":"040db777a2a312a23300a6253d7430fceca9870f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-07 23:40:29.000000000","message":"Patch Set 4:\n\nNeeds to be in alphabetic order in openstacksdk.config too.","accounts_in_message":[],"_revision_number":4},{"id":"57781c26aaf09f2eac8db5977694b551e9f85b2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-08 08:18:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4ad118d991dc91c5487a33769afda57bafb4761e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-08 08:19:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab47b9fad8be0b9dd2b6259724c1991265476476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 08:38:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2ae75bdfd344d5b2a6edf071ac29d1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/956a074636144b3ea5c82baa453f633d : SUCCESS in 6m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/82c8addbcbbf409aa5feb23bed8da406 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":5},{"id":"5a24483954da654863f60ab83103216664c45806","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-08 08:51:57.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a78d74d90d6d588bdad4c1fbcf632ed11a0a66b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-08 12:02:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8d15b44a9fc7ce7ec6fa5715b55641467e9de260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 12:03:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6c6b9388c084c5c722a347dda45dbba34b03c108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 12:15:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f655e563744f1ea160fbe239862f63\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/81724d6cb79c4cc89feffe168907b4d7 : SUCCESS in 6m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d619f4a0ecce4c2e9ec98749e7269cdd : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c9466fa58ea185bdea4b1f3710ba9cf5c8a69cdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 12:15:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"84e0eb6911ae10c3afb846fe6cb9fca664cc9ff8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 12:26:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b719045b88d408c9fa910daeb94ca36\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/5f5f47945ea14191b37ea849c3974d52 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":5},{"id":"50e4e9ed6fb8ce2a61a724e7383dfe3ab95f4a15","author":{"_account_id":19519,"name":"Abhishek Jaiswal","display_name":"jkabhishek (ajaiswal)","email":"abhishek.jaiswal@windriver.com","username":"aj355y","status":"Wind River System"},"date":"2023-08-08 12:31:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"dbee8648dc40a91aee514105eed19ef3908decb3","revisions":{"7e2dcf8d48e628020f43b065a634bc4b08a9e426":{"kind":"REWORK","_number":1,"created":"2023-08-02 21:04:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/890365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/890365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/890365/1"}}},"commit":{"parents":[{"commit":"ef82902d7277fb343f5066853014a3a5609190da","subject":"Allow cinder-core to act as osc/sdk service-core","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef82902d7277fb343f5066853014a3a5609190da"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-02 21:03:11.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-02 21:04:11.000000000","tz":0},"subject":"Allow designate-core as osc/sdk service-core","message":"Allow designate-core as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\ndesignate-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e2dcf8d48e628020f43b065a634bc4b08a9e426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e2dcf8d48e628020f43b065a634bc4b08a9e426"}]},"branch":"refs/heads/master"},"65b5e4cef38d769de6b786ec1d99372d047b8f3e":{"kind":"REWORK","_number":2,"created":"2023-08-03 14:43:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/890365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/890365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/890365/2"}}},"commit":{"parents":[{"commit":"c12307b5367299d0b77c058446c7d0f5efcebb58","subject":"Merge \"Change zuul tenant ansible version to 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c12307b5367299d0b77c058446c7d0f5efcebb58"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-02 21:03:11.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-03 14:43:47.000000000","tz":0},"subject":"Allow designate-core as osc/sdk service-core","message":"Allow designate-core as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\ndesignate-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/65b5e4cef38d769de6b786ec1d99372d047b8f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/65b5e4cef38d769de6b786ec1d99372d047b8f3e"}]},"branch":"refs/heads/master"},"4af6e35e82499b370ad18bc35f4e8f19169ec998":{"kind":"REWORK","_number":3,"created":"2023-08-07 22:42:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/890365/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/890365/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/890365/3"}}},"commit":{"parents":[{"commit":"c12307b5367299d0b77c058446c7d0f5efcebb58","subject":"Merge \"Change zuul tenant ansible version to 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c12307b5367299d0b77c058446c7d0f5efcebb58"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-02 21:03:11.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-07 22:42:43.000000000","tz":0},"subject":"Allow designate-core as osc/sdk service-core","message":"Allow designate-core as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\ndesignate-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4af6e35e82499b370ad18bc35f4e8f19169ec998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4af6e35e82499b370ad18bc35f4e8f19169ec998"}]},"branch":"refs/heads/master"},"b9781ed4a62bf7b58d84edf183cf095377121f7f":{"kind":"REWORK","_number":4,"created":"2023-08-07 22:48:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/890365/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/890365/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/890365/4"}}},"commit":{"parents":[{"commit":"988bec5dbca471ad518aa8c04df665d43de525ac","subject":"Allow masakari-core to act as osc/sdk service-core","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/988bec5dbca471ad518aa8c04df665d43de525ac"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-02 21:03:11.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-07 22:48:57.000000000","tz":0},"subject":"Allow designate-core as osc/sdk service-core","message":"Allow designate-core as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\ndesignate-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b9781ed4a62bf7b58d84edf183cf095377121f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b9781ed4a62bf7b58d84edf183cf095377121f7f"}]},"branch":"refs/heads/master"},"dbee8648dc40a91aee514105eed19ef3908decb3":{"kind":"REWORK","_number":5,"created":"2023-08-08 08:18:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/890365/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/890365/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/890365/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/890365/5"}}},"commit":{"parents":[{"commit":"0ed74c479d9f6ac9e5724d3ad3e838ebea189497","subject":"Merge \"Allow cyborg-core to act as osc/sdk service-core\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ed74c479d9f6ac9e5724d3ad3e838ebea189497"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-02 21:03:11.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-08-08 08:17:48.000000000","tz":120},"subject":"Allow designate-core as osc/sdk service-core","message":"Allow designate-core as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\ndesignate-core gerrit group -2..+2 votes (but not Workflow)\nto [access \"refs/heads*\"] for python-openstackclient\nand openstacksdk.\n\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Icc6fd8c1966d6acfdeb602bacebbeb4ceec011a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbee8648dc40a91aee514105eed19ef3908decb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbee8648dc40a91aee514105eed19ef3908decb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
