)]}'
{"id":"openstack%2Fproject-config~890374","triplet_id":"openstack%2Fproject-config~master~Id93a5f92c94dbe45508e89a5e37639f5cdee4071","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-08-04 02:58:58.000000000","reason":"Change was submitted"},"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2023-08-04 02:58:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id93a5f92c94dbe45508e89a5e37639f5cdee4071","subject":"Allow masakari-core to act as osc/sdk service-core","status":"MERGED","created":"2023-08-03 01:31:16.000000000","updated":"2023-08-04 03:05:43.000000000","submitted":"2023-08-04 02:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890374","meta_rev_id":"39fa56a32861496644c88b296698adf570c08941","_number":890374,"virtual_id_number":890374,"owner":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 02:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":2,"date":"2023-08-04 02:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2023-08-04 02:48:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 01:43:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-03 05:12:23.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-08-04 02:48:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"5a6111a6e48d6f3f2577fac68d22cb1f9e94689d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-08-03 01:31:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eee417249bb1e320ffb364bb0fb1874f301c1a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 01:43:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841aaa2855f249cd8497c45f13636c85\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9fb9ea6dda9d40279436b320516a4ba5 : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/50ac2c6554ef43af802953589db2dac5 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fcb03a97c915be8e2ee0a32f0287465f3a020fc8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-08-03 05:12:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03292c33e519fcce9a6b818494771229bcd65ec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"date":"2023-08-04 02:16:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1fcec3725f3756434cdf6ee2473d3a6972ecfba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 02:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/916d595113e242dca977136095e30e05\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea070bfa0b67422387d54a8e0c37ff91 : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bed3dabe1fec401bb8fba024c6f11b10 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0220578624e559c414cfa3068d7430737c25b3f2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-04 02:48:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nCarrying over Artem\u0027s earlier +1 for this rebase.","accounts_in_message":[],"_revision_number":2},{"id":"bfa496f16bcfe3291303d812da43890c3baacf36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 02:49:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1be5e39ca07fea6d32434239c84f0826cbf6aa6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 02:58:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1627adad548f4632a87c666ed8ae289e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/752d54c9612e47079ede407f533fe5fc : SUCCESS in 8m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5a573c3e2348451e9eac80a6a4d28903 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b8244bbc5c6fc4c60fa81da89d5e0eb57d7a1428","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 02:58:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"39fa56a32861496644c88b296698adf570c08941","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 03:05:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98c63a0a47ac44c1ac5ca8cfd75266ee\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/dd1ae960fe7a4e01ae99ed95f6d83fce : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"988bec5dbca471ad518aa8c04df665d43de525ac","revisions":{"f17fc29f278daa771569b12a5144ef7e0cdb9d20":{"kind":"REWORK","_number":1,"created":"2023-08-03 01:31:16.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/74/890374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/890374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/890374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/890374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/890374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/890374/1"}}},"commit":{"parents":[{"commit":"ef82902d7277fb343f5066853014a3a5609190da","subject":"Allow cinder-core to act as osc/sdk service-core","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef82902d7277fb343f5066853014a3a5609190da"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-08-03 01:19:54.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-08-03 01:20:09.000000000","tz":480},"subject":"Allow masakari-core to act as osc/sdk service-core","message":"Allow masakari-core to act as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\nmasakari-core gerrit group -2..+2 votes (but not Workflow)\nto [access refs/heads*] for python-openstackclient\nand openstacksdk.\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Id93a5f92c94dbe45508e89a5e37639f5cdee4071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f17fc29f278daa771569b12a5144ef7e0cdb9d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f17fc29f278daa771569b12a5144ef7e0cdb9d20"}]},"branch":"refs/heads/master"},"988bec5dbca471ad518aa8c04df665d43de525ac":{"kind":"REWORK","_number":2,"created":"2023-08-04 02:16:45.000000000","uploader":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"ref":"refs/changes/74/890374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/890374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/890374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/890374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/890374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/890374/2"}}},"commit":{"parents":[{"commit":"c12307b5367299d0b77c058446c7d0f5efcebb58","subject":"Merge \"Change zuul tenant ansible version to 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c12307b5367299d0b77c058446c7d0f5efcebb58"}]}],"author":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-08-03 01:19:54.000000000","tz":480},"committer":{"name":"suzhengwei","email":"suzhengwei@inspur.com","date":"2023-08-04 02:16:14.000000000","tz":480},"subject":"Allow masakari-core to act as osc/sdk service-core","message":"Allow masakari-core to act as osc/sdk service-core\n\nFollowing the Vancouver 2023 Forum discussion [0], give the\nmasakari-core gerrit group -2..+2 votes (but not Workflow)\nto [access refs/heads*] for python-openstackclient\nand openstacksdk.\n[0] https://etherpad.opendev.org/p/oscsdk-vancouver-forum-2023\n\nChange-Id: Id93a5f92c94dbe45508e89a5e37639f5cdee4071\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/988bec5dbca471ad518aa8c04df665d43de525ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/988bec5dbca471ad518aa8c04df665d43de525ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
