)]}'
{"id":"openstack%2Fproject-config~891566","triplet_id":"openstack%2Fproject-config~master~Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc","project":"openstack/project-config","branch":"master","topic":"app-intel-device-plugins","attention_set":{},"removed_from_attention_set":{"35620":{"account":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"last_update":"2023-08-16 14:29:10.000000000","reason":"Change was submitted"},"24777":{"account":{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},"last_update":"2023-08-16 14:29:10.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2023-08-16 13:32:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"28558":{"account":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"last_update":"2023-08-16 13:18:32.000000000","reason":"\u003cGERRIT_ACCOUNT_28558\u003e replied on the change","reason_account":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-08-16 12:14:09.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc","subject":"Add Intel Device Plugins app to StarlingX","status":"MERGED","created":"2023-08-16 07:16:00.000000000","updated":"2023-08-16 15:38:41.000000000","submitted":"2023-08-16 14:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891566-app-intel-device-plugins","meta_rev_id":"854a82d584aad6d521734a70aa122db7a80662fe","_number":891566,"virtual_id_number":891566,"owner":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-16 14:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"all":[{"value":1,"date":"2023-08-16 13:32:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-16 13:18:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":2,"date":"2023-08-16 12:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":1,"date":"2023-08-16 14:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 07:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 08:58:52.000000000","updated_by":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2023-08-16 08:58:52.000000000","updated_by":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-08-16 12:24:31.000000000","updated_by":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"reviewer":{"_account_id":24777,"name":"Steven Webster","email":"steven.webster@windriver.com","username":"swebster"},"state":"REVIEWER"},{"updated":"2023-08-16 12:24:31.000000000","updated_by":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"REVIEWER"}],"messages":[{"id":"e4bec6412a9bc9edf2eb26c172cfa6630b43e6c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"date":"2023-08-16 07:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c10ba0008c2d1eb84a4995200c5bfe6fc1fa810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"date":"2023-08-16 07:19:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a9a59f4f7349a4946608c3d9ac2d2c7f235cb008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 07:26:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9874724df214833aa663ec511a1a41b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d534bccedebe4f71909ff80e93eba47e : SUCCESS in 5m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5d6a42ad6093432e9302ff1f3a175b5e : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"197a480dae7f952417cf3d6383be7f726f8601ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"date":"2023-08-16 07:34:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7eb562e9173255997aa2b1435888ea98762b39e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 07:41:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/731c7ee68f2142219de68d83d81e475c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/daf4897b1cbd4a7091d81ae2f2333f29 : SUCCESS in 5m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7f48b72f6e7b4500a18fdaafc9c4ef95 : FAILURE in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"d1de0b2a0986c4373f4a931941ef7cf31b9c2b8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"date":"2023-08-16 07:44:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"991b01e075efacb66cd1e745123fa9c540c8ab0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 07:53:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78f7d10edbf64c6aa9cb29db49466fbc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae7a8075bc724be2968136cb817153e1 : SUCCESS in 5m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f8188edcb57845bc9eb50b3fa22250f3 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a44acfe906265ce5bba44037aa9736a7de526982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"date":"2023-08-16 08:42:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"dd6ab3979caa0b6f91d9ea6b46dc200bcb3250d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 08:56:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9829f37ebf2c46259c44b025a75b951f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/329a3fafb4484563a0fcd1a80c3658df : SUCCESS in 6m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5b92a7d5c5ae469e9af8c6d83c95cd83 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5b449e2cd63144f06b07ed0f43f7b84488a74833","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-16 12:14:09.000000000","message":"Patch Set 5: Code-Review+2\n\nAwaiting thumbs-up from Scott to merge, and then I\u0027ll add him as the initial member of the new core group once the change deploys successfully.","accounts_in_message":[],"_revision_number":5},{"id":"fbf861630f8cb1485b092581f0a73fa9028f8aec","author":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"date":"2023-08-16 13:18:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3621c9f594269c9999cbf6978870cc225b65bbb0","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2023-08-16 13:32:05.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8ad5c8196c47376b67d894c4e31f88c254d3104a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-16 14:14:17.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"70c7f10c911b6bfe919eb8bff3afb4a40e937167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 14:14:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9d2d71e0f38c187b5cd43e0de8ff8b3bbd3e07e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 14:29:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3211baa4ed28433cac79cb8c9e9a56aa\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d619285b1d9849de9ad5f0f32229bed7 : SUCCESS in 6m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/67d930dbe8344efaaeb09027f4976eb8 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":5},{"id":"2804b75064b9443e722a91279740b28a8c554634","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 14:29:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fdf4b9e6a4d5f8b94e9005bb1cbe4da13dc72fec","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 14:42:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b2c62782f304c71b10022d132597a27\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/3cdf869d22234cfb9a4abade8b2600e4 : SUCCESS in 6m 38s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/a077b54ba23b4e75b9ba9bfb6af22a08 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":5},{"id":"854a82d584aad6d521734a70aa122db7a80662fe","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-16 15:38:41.000000000","message":"Patch Set 5:\n\nI have added Scott as the initial member of the new starlingx-app-intel-device-plugins-core group now.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e817878affc5babd227fae3e1e11d4c8dd31f630","revisions":{"bb77ffbba81a1fad7b364276035808b00fd6af92":{"kind":"REWORK","_number":1,"created":"2023-08-16 07:16:00.000000000","uploader":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"ref":"refs/changes/66/891566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/891566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/891566/1"}}},"commit":{"parents":[{"commit":"d50921e66b0240c89064cd063ebadac844de6d32","subject":"Revert \"Revert \"Temporarily pause image uploads to rax-iad\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d50921e66b0240c89064cd063ebadac844de6d32"}]}],"author":{"name":"msheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:15:31.000000000","tz":-240},"committer":{"name":"msheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:15:31.000000000","tz":-240},"subject":"Add Intel Device Plugins app to StarlingX","message":"Add Intel Device Plugins app to StarlingX\n\nStory: 2010604\nTask: 48629\n\nChange-Id: Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc\nSigned-off-by: msheikh \u003cmdirshad.sheikh@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb77ffbba81a1fad7b364276035808b00fd6af92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb77ffbba81a1fad7b364276035808b00fd6af92"}]},"branch":"refs/heads/master"},"010fa28d5e91c2d89323f86c303b184b78f27d5e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-16 07:19:33.000000000","uploader":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"ref":"refs/changes/66/891566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/891566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/891566/2"}}},"commit":{"parents":[{"commit":"d50921e66b0240c89064cd063ebadac844de6d32","subject":"Revert \"Revert \"Temporarily pause image uploads to rax-iad\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d50921e66b0240c89064cd063ebadac844de6d32"}]}],"author":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:15:31.000000000","tz":-240},"committer":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:18:59.000000000","tz":-240},"subject":"Add Intel Device Plugins app to StarlingX","message":"Add Intel Device Plugins app to StarlingX\n\nStory: 2010604\nTask: 48629\n\nChange-Id: Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc\nSigned-off-by: Md Irshad Sheikh \u003cmdirshad.sheikh@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/010fa28d5e91c2d89323f86c303b184b78f27d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/010fa28d5e91c2d89323f86c303b184b78f27d5e"}]},"branch":"refs/heads/master"},"5da529d0b659f500539f021b2520deacc302c844":{"kind":"REWORK","_number":3,"created":"2023-08-16 07:34:26.000000000","uploader":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"ref":"refs/changes/66/891566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/891566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/891566/3"}}},"commit":{"parents":[{"commit":"d50921e66b0240c89064cd063ebadac844de6d32","subject":"Revert \"Revert \"Temporarily pause image uploads to rax-iad\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d50921e66b0240c89064cd063ebadac844de6d32"}]}],"author":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:15:31.000000000","tz":-240},"committer":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:34:18.000000000","tz":-240},"subject":"Add Intel Device Plugins app to StarlingX","message":"Add Intel Device Plugins app to StarlingX\n\nStory: 2010604\nTask: 48629\n\nChange-Id: Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc\nSigned-off-by: Md Irshad Sheikh \u003cmdirshad.sheikh@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5da529d0b659f500539f021b2520deacc302c844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5da529d0b659f500539f021b2520deacc302c844"}]},"branch":"refs/heads/master"},"f281b0bbf0af884519ee49ffce44e3415cf0f8e1":{"kind":"REWORK","_number":4,"created":"2023-08-16 07:44:07.000000000","uploader":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"ref":"refs/changes/66/891566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/891566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/891566/4"}}},"commit":{"parents":[{"commit":"d50921e66b0240c89064cd063ebadac844de6d32","subject":"Revert \"Revert \"Temporarily pause image uploads to rax-iad\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d50921e66b0240c89064cd063ebadac844de6d32"}]}],"author":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:15:31.000000000","tz":-240},"committer":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:44:03.000000000","tz":-240},"subject":"Add Intel Device Plugins app to StarlingX","message":"Add Intel Device Plugins app to StarlingX\n\nStory: 2010604\nTask: 48629\n\nChange-Id: Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc\nSigned-off-by: Md Irshad Sheikh \u003cmdirshad.sheikh@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f281b0bbf0af884519ee49ffce44e3415cf0f8e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f281b0bbf0af884519ee49ffce44e3415cf0f8e1"}]},"branch":"refs/heads/master"},"e817878affc5babd227fae3e1e11d4c8dd31f630":{"kind":"REWORK","_number":5,"created":"2023-08-16 08:42:47.000000000","uploader":{"_account_id":35620,"name":"Md Irshad Sheikh","display_name":"Md Irshad Sheikh","email":"MdIrshad.Sheikh@windriver.com","username":"msheikh","status":"WindRiver Systems"},"ref":"refs/changes/66/891566/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/891566/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/891566/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/891566/5"}}},"commit":{"parents":[{"commit":"d50921e66b0240c89064cd063ebadac844de6d32","subject":"Revert \"Revert \"Temporarily pause image uploads to rax-iad\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d50921e66b0240c89064cd063ebadac844de6d32"}]}],"author":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 07:15:31.000000000","tz":-240},"committer":{"name":"Md Irshad Sheikh","email":"mdirshad.sheikh@windriver.com","date":"2023-08-16 08:42:34.000000000","tz":-240},"subject":"Add Intel Device Plugins app to StarlingX","message":"Add Intel Device Plugins app to StarlingX\n\nStory: 2010604\nTask: 48629\n\nChange-Id: Ibbfabea9d11580854fe06cbff363ecdf1ed01bcc\nSigned-off-by: Md Irshad Sheikh \u003cmdirshad.sheikh@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e817878affc5babd227fae3e1e11d4c8dd31f630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e817878affc5babd227fae3e1e11d4c8dd31f630"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
